)]}'
{"id":"openstack%2Foslo.upgradecheck~690816","triplet_id":"openstack%2Foslo.upgradecheck~master~I2831ab17d943638062b709882c7a1cc481e46476","project":"openstack/oslo.upgradecheck","branch":"master","hashtags":[],"change_id":"I2831ab17d943638062b709882c7a1cc481e46476","subject":"tox: Keeping going with docs","status":"MERGED","created":"2019-10-24 06:10:16.000000000","updated":"2019-11-06 02:38:19.000000000","submitted":"2019-11-06 02:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690816-1573007833641-dd5696ed","meta_rev_id":"6bc8413f6bc76b9e5c51e0f2f06b3635957b2f09","_number":690816,"virtual_id_number":690816,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-10-25 16:08:48.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 02:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-11-06 02:27:34.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-10-26 13:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-11-06 02:27:34.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 09:24:28.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-10-26 13:28:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-06 02:27:34.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-06 02:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb6f0ae0440ee56b4701b74a5e79883d99b189fd","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-24 06:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8185cee7d616b2b9157b063bea37103967802ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 06:32:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784cad7b43ae459e8dd1370288613475 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eec72e73e224be5977210e32b11ab8a : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43f8063dbc014dee8ff87298ab3f1f6d : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b063f13c828547619351638c5954cfaa : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/162d2f8f3d994ac3ae0d5d9229c0e77f : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bbaf1afa1b584fa5bad6a155b87698ce : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d26cd8532ebd4624a68393d585b00d3d : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e19af4a3d2c415fbe601c5cb19767e7 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"230a8b18bc118517bd01010cfdf7f77ec3bf1212","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-24 07:14:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc8ff29112231948f22591431e1f1644fef7e228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 07:37:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8479a417b6864beba9d1b936f8a0dc75 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b99ab6fd7e643c69d2075a325ce5a34 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e87212a9ce642b68f3d6f10920c2f38 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e66db539e05642ada1268360994df614 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eebc8fe46d54a469f1d4fc7fdfc1528 : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c26f4a423aba41c0891012d14cc3c9e7 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f759ed969f344fe080c16921a585b0d3 : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d761b0fb5d5240c18001123e3fd4aa6c : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5e0a0fc4847b745486360607a4668a608e5a00ca","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-24 09:24:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"115281cc4864c1052af2162f9971aed543c766e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-25 15:55:32.000000000","message":"Patch Set 2: Code-Review-1\n\nYou also need to update:\n\nhttps://github.com/openstack/oslo.upgradecheck/blob/226d9b14d3e18e13887326da8a9af99325e6fa07/lower-constraints.txt#L8","accounts_in_message":[],"_revision_number":2},{"id":"0ae92289ad36fd71bcffed6b9124f40c2b047927","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-25 16:07:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3fa2a2c8a7d015c9af64ba3253c179ae342ff407","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-25 16:08:48.000000000","message":"Patch Set 3:\n\n\u003e You also need to update:\n \u003e \n \u003e https://github.com/openstack/oslo.upgradecheck/blob/226d9b14d3e18e13887326da8a9af99325e6fa07/lower-constraints.txt#L8\n\ndone, thanks for your remind","accounts_in_message":[],"_revision_number":3},{"id":"389be573f3a0e698a9bcebbd1c73725477114f0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:12:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f2f681e8df48c68f1dddc9fbdd958a : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3eeddea39549c3a201aa291dbbfc50 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e368942d45754825915f89e3bf4fdae4 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1b80b9c01d43358b78691ba6655c4e : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af2b495d4dff480dae124063f43fbbb2 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dacfd2148284f79b4196eef13d8447f : SUCCESS in 1m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/813c5f1ff82e4f6f9d33952f8e683cbf : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4b22a8b3b274983b37fd6050f2e1122 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":3},{"id":"3bdb003f8478bab1985012f3c5b586c375a75eea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-26 13:28:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d80de5d99f33f054184d6289e1e8a563d651a19d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-06 02:27:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7352bb9da31ab96aa88b3de1a935c69818961357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 02:27:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ef7adf309845acd7df61419cbd96e722eccba96f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 02:37:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2386750305e444c866a17bd1a4e8755 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c0eba381c645869106b877560299f8 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc271df593d24cfa87a0c2157f809178 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5907cce66094d079cf0a54d382a8eeb : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da25fc9333e44644b38285dbceec2670 : SUCCESS in 3m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d6dd65e4c05454c987e391b0a430a7e : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18f90e7e0033435aa4bf0100947df43a : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b004bf026b14488b963fc7fa84b9b413 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"70bde72db60b38e529b023f03c8335fdb201e90b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 02:37:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6bc8413f6bc76b9e5c51e0f2f06b3635957b2f09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 02:38:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8650ccb13be44075977a41c71982e11f : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c39d76e6a84136a7ea6c58cfcd65de : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9e70fdadee5ce346d4b8ad21be8192400e1bf3ae","revisions":{"53159f333ad73cc6ad3907e5cbc4c5baf0613654":{"kind":"REWORK","_number":1,"created":"2019-10-24 06:10:16.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/16/690816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/16/690816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/1"}}},"commit":{"parents":[{"commit":"9765b139136c26c29fca4c98d9a76ab77ee10ad7","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9765b139136c26c29fca4c98d9a76ab77ee10ad7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 06:09:59.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 06:09:59.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I2831ab17d943638062b709882c7a1cc481e46476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/53159f333ad73cc6ad3907e5cbc4c5baf0613654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/53159f333ad73cc6ad3907e5cbc4c5baf0613654"}]},"branch":"refs/heads/master"},"86d7f2d0bfe53e5824ce2a0c588f944e403a0f51":{"kind":"REWORK","_number":2,"created":"2019-10-24 07:14:34.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/16/690816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/16/690816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/2"}}},"commit":{"parents":[{"commit":"9765b139136c26c29fca4c98d9a76ab77ee10ad7","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9765b139136c26c29fca4c98d9a76ab77ee10ad7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 06:09:59.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 07:14:03.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nSphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: I2831ab17d943638062b709882c7a1cc481e46476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/86d7f2d0bfe53e5824ce2a0c588f944e403a0f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/86d7f2d0bfe53e5824ce2a0c588f944e403a0f51"}]},"branch":"refs/heads/master"},"9e70fdadee5ce346d4b8ad21be8192400e1bf3ae":{"kind":"REWORK","_number":3,"created":"2019-10-25 16:07:31.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/16/690816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/16/690816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/16/690816/3"}}},"commit":{"parents":[{"commit":"9765b139136c26c29fca4c98d9a76ab77ee10ad7","subject":"Merge \"Update master for stable/train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9765b139136c26c29fca4c98d9a76ab77ee10ad7"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-24 06:09:59.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-10-25 16:07:16.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nSphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: I2831ab17d943638062b709882c7a1cc481e46476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9e70fdadee5ce346d4b8ad21be8192400e1bf3ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9e70fdadee5ce346d4b8ad21be8192400e1bf3ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
