)]}'
{"id":"openstack%2Foslo.upgradecheck~700146","triplet_id":"openstack%2Foslo.upgradecheck~master~I7c672f05753886e234430c760f1afc1ed01ff2a3","project":"openstack/oslo.upgradecheck","branch":"master","topic":"base_python","hashtags":[],"change_id":"I7c672f05753886e234430c760f1afc1ed01ff2a3","subject":"tox: Trivial cleanup","status":"MERGED","created":"2019-12-20 10:16:08.000000000","updated":"2020-01-30 12:30:10.000000000","submitted":"2020-01-30 12:25:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700146-1580387143458-04bc3352","meta_rev_id":"aeb6e92dd41119bb19c2d3496cd81a13704b42ac","_number":700146,"virtual_id_number":700146,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 12:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2020-01-30 12:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-01-30 12:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-01-30 12:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 06:12:37.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-01-30 11:23:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 12:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87cf227a545d11e8665182e66f3ea896099e0f5c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 10:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be7f1a719c98e9e82841c188852bae2146eec271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:24:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc4e34328d048529e829cc42faddf53 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cabc928165f442daaef8a026e0e633a9 : SUCCESS in 3m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba97954b187443199e8eecbe7ff7ff46 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24d5535d7d1f4b6b8bc5637ca2b69085 : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de3b66acb8bf4fa8bed59bd878d945aa : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d30499994ac34f13a64e76b89ab6b08e : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d84a8353a49842ba989838fa89f83226 : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"559ea75884180c2984e2639714233e58aabd59e8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 15:02:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"815bc866e648d4965a93359cf74ec318a9cf8e9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 15:11:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533532ac3481408db4ff758b9dfb8869 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bf8f3c79e34e2cb360ed9fee58c882 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a5aa156bcf04430a01362742772d06b : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c86dc3bcbf848ec8270cc7444e061bb : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0508618ca56f40208a841871cff7fca7 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/841e384ad0ed418285e4e8f9b803abf2 : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd6801bc182b40e2924a3d061098aadf : FAILURE in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ef6ef3ab4686e027b43001f02490213a1dcc1801","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:41:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de8f8f7f4f60c32e068c200f699bcde2aefb3983","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:41:15.000000000","message":"Patch Set 1:\n\nLooks like lower-constraints needs a fix first too","accounts_in_message":[],"_revision_number":1},{"id":"197124b31c55a4a96d16746aeb45184924ab910f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-08 19:13:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f8e8631265daad29701ec81c6c98ab3fcc6fb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 19:25:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbed106c770642dc97d7431a632d9be4 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a844b6fab8747b3aa57979bc2389a6e : FAILURE in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a89019005c0045098732a5b1749882b3 : FAILURE in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8dfd76d36244f6bab5f225762f59835 : FAILURE in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17370feec09a40efa2f332fe12299553 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e6f88a323454cc68e47498560326397 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f15bb6753e6474f8eec89e0626df32e : FAILURE in 10m 38s","accounts_in_message":[],"_revision_number":2},{"id":"aaec31b94f184c3e371de214c748b48416ea4594","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-14 11:55:39.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e3a10a1d48735d76136d35ea26e1f3b410c9d5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:08:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff6c24ef2074e4086b8eb95ec937686 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7209695aa1bf448a924218d0df35a8c1 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0b981dd742f42258720bc6c67921448 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a1f1d097c1b44948f729a499de39825 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae9416e8df1443039dd43358ac0166ea : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0aa9d05d987410cb29529e5254600a0 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8e07b6ae7f1453d9cb9404c4b454645 : FAILURE in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c47a46baf3b4b2fd5575581c1dc3291faf8d9d0d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-15 08:56:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e266feab21eac5195d0bd75ffdcb2e7f3f7bf1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 09:05:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147ee13c620f428997d0fe0101a2889c : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12fbf73aaf94520933e26009560ab71 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/765941156c01477991ec673fdb98fe82 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b349bf2a7f754ffdbbf437d6f171ac87 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a01474dc516c4cdabcb123b650b9c1be : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/290f1c299ea441668e9239924ba2cd68 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba86a0770b8142bda74ca6da9cd5cd7b : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"951997b314b527b6f63355a254b3f8c40acd0458","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-16 12:29:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f3615406c7eaa91a0b8ba7621267abded2db6621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:39:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85f27e59d2a4f5388172b8cbf33d67b : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c0eae3ecbf4c91a19ace2ca4a9286f : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/723c371f85ec4ffebf50e4d89128fa8c : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5267cba4fca43dfa16035056fbf639c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/daf0c7221fd942d18b03bec670f61d7a : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e73b8879608f402c8f801354ffd2da83 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09460095d20644529e978d64ad198d3f : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"575cdf6d6cf62d1f59a2634c3d7a49b60e128e4e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-16 12:39:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0dc5c50234a5cca2d6c33adb640b8c4de1f90503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:48:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4c986b981846f4828d1a81d3bc7109 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b518924bc347d78798884057606b5a : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d7e569a56d14863915e198ef6b6631e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/932dd91e66bd4e78986c9ff329fe3479 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd65f73e25574f208be662749b71f896 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a22e3544e294defa692ee20be74c5de : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190736edc9054d089b37e27d42f2485f : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"49ae9843eec569988c46bfd17986231e02d5a2fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 16:25:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"60b6bbff866982b7681392482c624b650d8a407f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 16:35:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c00ad4806f45968c88f19dc05c2d78 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c2f8d8e4ac4950884c540549729b46 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95bbda4d227d481aa1d8db471e5245fc : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/504851f0a1864f308aa588c0a47defb8 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef4da8008b1c4bd7b5163c12ae03f3eb : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aea59922d99941aa938db8203078cce4 : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/732b6e39f34f43fba4683a9d43e6be1d : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"97ea58e6b181c7760539ba33c0b6276fbc047727","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-30 06:12:37.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good.","accounts_in_message":[],"_revision_number":4},{"id":"58e098cbee29f12e931e8433bf440da4f887a32a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 11:23:04.000000000","message":"Patch Set 4: Workflow+1\n\nThis is obviously correct and bnemec has other things on his plate at the moment. Let\u0027s keep this moving.","accounts_in_message":[],"_revision_number":4},{"id":"59c2e3a8bee0f6e3f20af539307b949caf344a92","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:01:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6116d4b4910fe875e4b1ac87c3cd6aaf98a8bd3b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:01:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f461f5ef2f7934ab4c80570c39df239aa8b0877a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:12:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"840460ee9bdc2c747030b3a55da24d69c80cb331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:12:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e117436f14264c10a3de312281a6c132 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037be8ce181541a7a5e744728e05a6d3 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6ec9f114ea14ae1a9987618cce68c25 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c581ca2f354f43a833448d98ce3f37 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/515c142db5a34809a284cad19ab12b01 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4088ea3ec5d47019c6c3bf83f8e128a : SUCCESS in 4m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3180d4731c4556ac1d4dd52fb207cc : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/923542fc29a14e50b29dfdfc4c8c4092 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":6},{"id":"9d566218a5c69b5ba3cf5ffbd8ab127b8bbd01c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:25:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d8c34684744f46839f58ed67ba1f16 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b47ac909ddd4135bae4cba733822a95 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/771addec6a4c4235ba2406101b5efd0f : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65358d7c90014dfb85f621e662b8bd16 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c6c4708fe90489cae5b734e55ba4fc4 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c807d0a09894fbead7e070d267b518c : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82d5b7f2f69e4d4f882026c11072dddc : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":6},{"id":"801b5a41896a7751336253e76f41f9956f50ed7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:25:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"bde94a38a334bb322c35f50dda1fb0c99bff42f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:30:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1ed2064a75406eb41cb873ad0a41df : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1be10a4c7d5470596026a47ee8c3c6c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2837bff516eb283e43ec0795d285b9485dbc4e53","revisions":{"9672b7803764fe1bb76502aa006667825c421d11":{"kind":"REWORK","_number":1,"created":"2019-12-20 10:16:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/700146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/1"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:16:05.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9672b7803764fe1bb76502aa006667825c421d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9672b7803764fe1bb76502aa006667825c421d11"}]},"branch":"refs/heads/master"},"88d59b11cfe4229135e9ffcc73a8a4d4ec0b58da":{"kind":"REWORK","_number":2,"created":"2020-01-08 19:13:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/700146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/2"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-08 19:13:43.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/88d59b11cfe4229135e9ffcc73a8a4d4ec0b58da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/88d59b11cfe4229135e9ffcc73a8a4d4ec0b58da"}]},"branch":"refs/heads/master"},"1fc931149f28e0a091985c1d3ad5bcaf24edf00e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-16 12:39:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/700146/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/3"}}},"commit":{"parents":[{"commit":"c9b5177c6aff747f95f54a0b0eb054cc7fd4b0bf","subject":"Bump Babel to 2.3.4 on lower-constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c9b5177c6aff747f95f54a0b0eb054cc7fd4b0bf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:39:45.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/1fc931149f28e0a091985c1d3ad5bcaf24edf00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/1fc931149f28e0a091985c1d3ad5bcaf24edf00e"}]},"branch":"refs/heads/master"},"b16151a8fd4ae405efd9228f45d2a878bcba811d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-17 16:25:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/700146/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/4"}}},"commit":{"parents":[{"commit":"cd8dfa982ca85f9f186b4fd086ae4170c714ca74","subject":"Bump Babel to 2.3.4 on lower-constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cd8dfa982ca85f9f186b4fd086ae4170c714ca74"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-17 16:25:02.000000000","tz":60},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/b16151a8fd4ae405efd9228f45d2a878bcba811d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/b16151a8fd4ae405efd9228f45d2a878bcba811d"}]},"branch":"refs/heads/master"},"3731623056777863bf7339c7a96cc6c03eb2c940":{"kind":"NO_CHANGE","_number":5,"created":"2020-01-30 12:01:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/700146/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/5"}}},"commit":{"parents":[{"commit":"03837f2b62b5997137c4129fb2d3283cf1b20100","subject":"Bump Babel to 2.3.4 on lower-constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/03837f2b62b5997137c4129fb2d3283cf1b20100"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-30 12:01:08.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/3731623056777863bf7339c7a96cc6c03eb2c940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/3731623056777863bf7339c7a96cc6c03eb2c940"}]},"branch":"refs/heads/master"},"2837bff516eb283e43ec0795d285b9485dbc4e53":{"kind":"NO_CHANGE","_number":6,"created":"2020-01-30 12:01:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/700146/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/46/700146/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/46/700146/6"}}},"commit":{"parents":[{"commit":"1024d00f10e52273c6c021f221191e9b1a9447ae","subject":"Bump Babel to 2.3.4 on lower-constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/1024d00f10e52273c6c021f221191e9b1a9447ae"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-12-20 10:15:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-01-30 12:01:47.000000000","tz":0},"subject":"tox: Trivial cleanup","message":"tox: Trivial cleanup\n\nMove \u0027basepython\u0027 to the top-level \u0027testenv\u0027.\nUse the default \u0027install_command\u0027\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nChange-Id: I7c672f05753886e234430c760f1afc1ed01ff2a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/2837bff516eb283e43ec0795d285b9485dbc4e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/2837bff516eb283e43ec0795d285b9485dbc4e53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
