)]}'
{"id":"openstack%2Foslo.upgradecheck~702859","triplet_id":"openstack%2Foslo.upgradecheck~master~Icd6095a373188f5ef152ec2f1748aa7f826ad8a8","project":"openstack/oslo.upgradecheck","branch":"master","topic":"base_python","hashtags":[],"change_id":"Icd6095a373188f5ef152ec2f1748aa7f826ad8a8","subject":"Bump Babel to 2.3.4 on lower-constraint","status":"MERGED","created":"2020-01-16 12:39:57.000000000","updated":"2020-01-30 12:28:39.000000000","submitted":"2020-01-30 12:25:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702859-1580387142848-04bc3352","meta_rev_id":"68de846def2eab158ed0dded351ad64df91488da","_number":702859,"virtual_id_number":702859,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 12:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-01-30 12:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-01-30 12:01:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 12:01:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-30 12:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39373a2d862d56aa5c05fa43025690bc6b86df62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-16 12:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1ec49b0f12f387fba87c8670a2e722384ddb7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 12:53:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00977315e96c4ddfb0177f37ab24c960 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819184f1834c4450ad89b033de7e5478 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e098c0a3350b468cb42fdd52cf5b75c6 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e986087c7bb544cd9355e963b56e86c4 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a285758b0fb547fa95548e39a2021e1f : SUCCESS in 5m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61b3c05f792a4649ad380472518ca8c8 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c521a4bdadb640d0849d07d1fed0d47f : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d47ace1fbea008c45d3981e1cbeff93678a4d069","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 16:25:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f805bcdbd2c4444bfb197ce43f005f4a985d11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 16:35:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42b8f7c75484d0f80999245fe6b9eb8 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a16edc60bea40abb42f35aef5ffb668 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/171556df620d405e861b80077363cbed : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/494f91d0b71143c8a16b0158ee47ddcf : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e57c61989a2e4e1aa7aff778cfc006c2 : SUCCESS in 4m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99b708fe8e934089b55e3725d1e29774 : SUCCESS in 2m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/830812c424d346cb81eefdfeb99a575a : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"45b124956fc8ef583570df83b3a1afb3b0dea27b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 11:23:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"037ce0120e8400695bc590451d331e189cebfdd9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-30 11:56:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb3bf7e05c96a2083410af3d0ba4560047c2532f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-30 12:00:37.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7545d879e0478d3cd92c0a60d590d3cbfc66e382","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:01:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f53c9f08d491416d1faadbd22c0f203a84f672f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-30 12:01:23.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f20ca3c11a430709b75c6ecc63ab590c7d7c9fa2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-30 12:01:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b5c95d8d65cc682961ad2ace34f6643028eecfb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:12:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad61bba61ed34e019239ddfd57c45ffb : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1923c85aaa3413fac0c2cca1136bcc2 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6adabce3fb045b7924e452315dc590b : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be6914509aa1435186aa831fe33258dc : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bdb5a627d184f5fb6c667255cf518ea : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76e112b13351472d984d50785e91efa9 : SUCCESS in 5m 45s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/02e87bfdc5454657b19e528017b3c348 : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4e3c12f19a24f75b895847a5d0c483c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"a10bc7d799472c942fe3ec23b1f85afe41c5568a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:12:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9f0eac1aba6aeb42bc16684b55cc3d1c3e5d915e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:25:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cd3b95f27962950387853b707ddf8fc044ae3746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:25:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512b9bb66f01417aacdf724c152e75b7 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbb01298ac2f49d8976c785e08cd2216 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df1f6679bb544a249188ffe026a814df : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/988c55cfd8744748b345fc01914f14e7 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d993ffd664d402bbcf46a2d21378859 : SUCCESS in 3m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/452e9c2747534f2b82ba9fb6bdccb365 : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07bd874c7f094aaf881f4443afb044ea : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":4},{"id":"f5919c25da71b84588d97707a2a367c048a42fa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:28:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771424b76db04442a0961bb106723865 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1024d00f10e52273c6c021f221191e9b1a9447ae","revisions":{"c9b5177c6aff747f95f54a0b0eb054cc7fd4b0bf":{"kind":"REWORK","_number":1,"created":"2020-01-16 12:39:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/702859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/59/702859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/1"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:38:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:39:45.000000000","tz":60},"subject":"Bump Babel to 2.3.4 on lower-constraint","message":"Bump Babel to 2.3.4 on lower-constraint\n\nChange-Id: Icd6095a373188f5ef152ec2f1748aa7f826ad8a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c9b5177c6aff747f95f54a0b0eb054cc7fd4b0bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c9b5177c6aff747f95f54a0b0eb054cc7fd4b0bf"}]},"branch":"refs/heads/master"},"cd8dfa982ca85f9f186b4fd086ae4170c714ca74":{"kind":"REWORK","_number":2,"created":"2020-01-17 16:25:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/702859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/59/702859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/2"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:38:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-17 16:25:02.000000000","tz":60},"subject":"Bump Babel to 2.3.4 on lower-constraint","message":"Bump Babel to 2.3.4 on lower-constraint\n\nChange-Id: Icd6095a373188f5ef152ec2f1748aa7f826ad8a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cd8dfa982ca85f9f186b4fd086ae4170c714ca74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cd8dfa982ca85f9f186b4fd086ae4170c714ca74"}]},"branch":"refs/heads/master"},"03837f2b62b5997137c4129fb2d3283cf1b20100":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-30 12:00:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/702859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/59/702859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/3"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:38:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-30 11:59:29.000000000","tz":60},"subject":"Bump Babel to 2.3.4 on lower-constraint","message":"Bump Babel to 2.3.4 on lower-constraint\n\nCI jobs complains about lower constraint who was not aligned so moving\nBabel from 1.3 to 2.3.4\n\nChange-Id: Icd6095a373188f5ef152ec2f1748aa7f826ad8a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/03837f2b62b5997137c4129fb2d3283cf1b20100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/03837f2b62b5997137c4129fb2d3283cf1b20100"}]},"branch":"refs/heads/master"},"1024d00f10e52273c6c021f221191e9b1a9447ae":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-30 12:01:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/702859/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/59/702859/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/59/702859/4"}}},"commit":{"parents":[{"commit":"9eb8518b84a76b9d5935a744387ba3c6333e5fba","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9eb8518b84a76b9d5935a744387ba3c6333e5fba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-16 12:38:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-30 12:00:53.000000000","tz":60},"subject":"Bump Babel to 2.3.4 on lower-constraint","message":"Bump Babel to 2.3.4 on lower-constraint\n\nCI jobs complained about lower constraint who was not aligned so moving\nBabel from 1.3 to 2.3.4\n\nChange-Id: Icd6095a373188f5ef152ec2f1748aa7f826ad8a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/1024d00f10e52273c6c021f221191e9b1a9447ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/1024d00f10e52273c6c021f221191e9b1a9447ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
