)]}'
{"id":"openstack%2Foslo.upgradecheck~706933","triplet_id":"openstack%2Foslo.upgradecheck~master~I0de4e6a064fb8008f9cff61b96875909041dd14b","project":"openstack/oslo.upgradecheck","branch":"master","topic":"remove-outdated-headers","hashtags":[],"change_id":"I0de4e6a064fb8008f9cff61b96875909041dd14b","subject":"remove outdated header","status":"MERGED","created":"2020-02-10 17:37:35.000000000","updated":"2020-02-13 10:47:38.000000000","submitted":"2020-02-13 10:40:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706933-1581590459471-027efaee","meta_rev_id":"0b485d026c6cc0302e7c963a74be67e5ae30fd4d","_number":706933,"virtual_id_number":706933,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-02-12 10:20:29.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 10:40:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-13 10:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-13 10:28:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 14:43:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 10:28:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-13 10:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e66b6fce5d9ada80321ba183a8ef17b4af4a19f1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-10 17:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a03c7858fa4cbfdee8ca40bd1aef1676ad99558d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:58:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e15e03eeb2de46048b2173bebcf8b80b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9adfc3866a4944539a084ef8f7cfb38a : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfebaeb8e9e42d992f6f67fd686de29 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ee13bec3ec849539d14dde0d7641836 : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/877fd120623844c7a199163a1afcdd4e : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62440a3db4ce4ce3841d6766ed342c6d : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b72fd3853f449eeab7a4543a9e8854b : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c65e34373ee47ec94f38f8d51ba10cb : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"96a1cc0ec3639debd4ea17d07bd9f65a2d986fd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 14:43:53.000000000","message":"Patch Set 1:\n\nNeeds https://review.opendev.org/#/c/705722/","accounts_in_message":[],"_revision_number":1},{"id":"99d80c3f2dfa743361c77fdf3499a6ad8338f67a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 17:47:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"68d2e891357b2384a314ad017ceb5b91e6843f89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 17:48:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dc2bfa678c3c19a3f1c18e2f0d9e26f08d75908a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 18:30:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0ebf5588d8094bb0bacff52eb2a61ff2 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1869aaecd22140e3933b825bda3a6055 : RETRY_LIMIT in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d942ce84332147c1ba576037726f24a4 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7db35ea4a1b4bc9a682a5a0593a02e5 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee95c06ed3814d7bb54cceb2e8908f80 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a79c1f4fb44d466eaa2de51dc9dbded1 : RETRY_LIMIT in 2m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e84eb457a828402d8d02c6d79cf3d8cb : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"709a0c63e3ff8261dd432e78a7418303778ff295","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-12 10:20:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf52570e6253379ed302cca3cef8374872f9e8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 10:31:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a80686a90fff4660b3c76ce32520cfbf : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6720cdf92dad48d9b0359f8d78121ede : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fb0777f2ed4c26983835cec7a1e225 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89c79f24ad3b4d49a44f7033958a0953 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/891f410cda984582be5e528312c083b1 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cbc1f81a1064bf58acdb4c4d2984af7 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0ca27d78419473a8d52bc8a0364e795 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b5e539d901caa6788ca06aeed69dfd58ec2b5506","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-13 10:28:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0816519c6c5f255782c7719c6426af0aa0e1bec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:29:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"442fe6c29247751b8efcc48b311c1bbbd507433c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:40:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"728c28c3fe1ff2f179e85f80b67a51023cc4dbf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:40:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f81e153e20054c2d83c919bf9c115921 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020f8248dc494687ba57114cbee72af7 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ddc1ce470dd4f0fb3880e9f1a03b7ee : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b592a986d2b24093941d816b6c967031 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9acfe7456a09437c8ccbc413d03f4432 : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0305d82d0aef428baddbdd4d93c69859 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"838f58ecfae2cc1c41a86918dfb436b8d4ec2025","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:47:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a17d8c840604080890a68e1f0c30374 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f391092a72ac50c564e59d6db95c216781de8576","revisions":{"cadd8e14756c29ed34cf309c9acf7f574579aaea":{"kind":"REWORK","_number":1,"created":"2020-02-10 17:37:35.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/706933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/33/706933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/1"}}},"commit":{"parents":[{"commit":"790d8e8391a918e338302e02410fd41fa2b7e508","subject":"Merge \"tox: Trivial cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/790d8e8391a918e338302e02410fd41fa2b7e508"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:37:30.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:37:30.000000000","tz":60},"subject":"remove outdated header","message":"remove outdated header\n\nChange-Id: I0de4e6a064fb8008f9cff61b96875909041dd14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cadd8e14756c29ed34cf309c9acf7f574579aaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cadd8e14756c29ed34cf309c9acf7f574579aaea"}]},"branch":"refs/heads/master"},"343fd2a20ee3e082f7e75812dbdf1163abf8d396":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-11 17:47:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/706933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/33/706933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/2"}}},"commit":{"parents":[{"commit":"0d84429800639596a060ac7f16a5d54c9eef6799","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/0d84429800639596a060ac7f16a5d54c9eef6799"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:37:30.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-11 17:47:32.000000000","tz":60},"subject":"remove outdated header","message":"remove outdated header\n\nChange-Id: I0de4e6a064fb8008f9cff61b96875909041dd14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/343fd2a20ee3e082f7e75812dbdf1163abf8d396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/343fd2a20ee3e082f7e75812dbdf1163abf8d396"}]},"branch":"refs/heads/master"},"f391092a72ac50c564e59d6db95c216781de8576":{"kind":"NO_CHANGE","_number":3,"created":"2020-02-11 17:48:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/706933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/33/706933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/706933/3"}}},"commit":{"parents":[{"commit":"0d84429800639596a060ac7f16a5d54c9eef6799","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/0d84429800639596a060ac7f16a5d54c9eef6799"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-10 17:37:30.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-11 17:48:10.000000000","tz":60},"subject":"remove outdated header","message":"remove outdated header\n\nChange-Id: I0de4e6a064fb8008f9cff61b96875909041dd14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/f391092a72ac50c564e59d6db95c216781de8576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/f391092a72ac50c564e59d6db95c216781de8576"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
