)]}'
{"id":"openstack%2Foslo.upgradecheck~719869","triplet_id":"openstack%2Foslo.upgradecheck~stable%2Fussuri~I3ef3bffe0d82fa1476714ea579d92a6d8301bbf6","project":"openstack/oslo.upgradecheck","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I3ef3bffe0d82fa1476714ea579d92a6d8301bbf6","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-14 10:17:52.000000000","updated":"2020-04-14 11:30:09.000000000","submitted":"2020-04-14 11:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719869-1586863642350-a663f06a","meta_rev_id":"e08e22ce6337f85b2d74274f7796b862c9f1d634","_number":719869,"virtual_id_number":719869,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 11:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-14 10:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-14 10:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-14 10:49:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 10:49:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-14 10:53:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-14 11:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a5f83caaa1756258b6d1c930b15ed66883b76d5","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-14 10:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1678b70d2dd297d7d799de2bcef098ac311e5051","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 10:49:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45a07f9dec8b4d3a3222c4ca073681c81e9451a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 10:53:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31fbf1af697ae902deb990d1570101fb49b59fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:15:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f72b2cbfd8428f8890f5d65b00e640 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c27c26a2804ce085736ee3e6970afc : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59deb1e5649d49b8a73c369ada92ac4b : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e79dc54dba2944b8b8daa634ca40f777 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8888344f1e8c41c390073a71923b0cfa : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58ae1e399f9642bbb19d9dbb3b5cd1f7 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cf6eecc932e210f33dbca255d779172970a5b4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fca710badc3850f22a39c8a401e69c7bc8c6d4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:27:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1562d508184353abfb3ba26ec076a0 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2077171025444cef92e4a406be0da147 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cce06fdab3d433e805b546df050ff75 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/148a8f3d96c549a2b53e962eadf0343f : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bbee5f7a68e4c2cb0004cfa353f7ab0 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92541230b66640f3877e6c967dd50af9 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"562f75f314297934b0a4e39c53c4a6e231327bf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:27:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc6a2083ccd00a8691ffa56ba60e39861b4a37e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:30:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a4c3008f0844eaaa94408c6e45bdc9 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ebd18ab6d8c41e7824566081adfd660 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"910cf672b90cceeee2fa744f892ab0a0ac3ed6f5","revisions":{"910cf672b90cceeee2fa744f892ab0a0ac3ed6f5":{"kind":"REWORK","_number":1,"created":"2020-04-14 10:17:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/719869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/69/719869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/719869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/719869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/719869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/719869/1"}}},"commit":{"parents":[{"commit":"c7bfccec8c2ab7293ac837a6ffbebd8e216c0163","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c7bfccec8c2ab7293ac837a6ffbebd8e216c0163"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:17:50.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:17:50.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I3ef3bffe0d82fa1476714ea579d92a6d8301bbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/910cf672b90cceeee2fa744f892ab0a0ac3ed6f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/910cf672b90cceeee2fa744f892ab0a0ac3ed6f5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
