)]}'
{"id":"openstack%2Foslo.upgradecheck~742147","triplet_id":"openstack%2Foslo.upgradecheck~master~I8d5290d62b12a7850fc932c3778cc37951dd67ed","project":"openstack/oslo.upgradecheck","branch":"master","topic":"oslo-pre-commit","hashtags":[],"change_id":"I8d5290d62b12a7850fc932c3778cc37951dd67ed","subject":"Adding pre-commit","status":"MERGED","created":"2020-07-21 11:37:34.000000000","updated":"2020-09-22 11:26:08.000000000","submitted":"2020-09-22 11:23:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742147-1600773826944-e4987785","meta_rev_id":"95dd0743a7cfa8603565d19f05d5262470894e7a","_number":742147,"virtual_id_number":742147,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 11:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2020-09-03 13:15:47.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2020-09-22 11:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2020-09-22 11:13:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 13:15:47.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-09-22 11:13:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-22 11:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95b60ccb6e5d1a648df7a07093aa656e379dc57f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 11:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b6f5e8aec2dfd7ba5407c10c4b04d900865c5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:50:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45054aafa1f04977b37d4685e7457e2b : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0a49f43703423faa6c480549518722 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/782e18cefe60446c9b919aa5bbeb0d17 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f126817775d4ba28e9ecc60a724c9e9 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70bacb06d439499f8a89e5a5b8935b35 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c8ee5cb3e144a02a8c3d63a5d0f66de : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2d96c2aad7cb2765cece037e4a7e4c4a11055530","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:43:50.000000000","message":"Patch Set 1: Code-Review-2\n\nWe should hold on in these patches for a while and observe oslo.cache for some time before spreading the change around all Oslo.","accounts_in_message":[],"_revision_number":1},{"id":"98664ef08acce23a22a9aa7b04e77cfa23820a15","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-03 12:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1c67b172fe4e4e6018dbb360b91e63ea28277b6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-03 12:11:09.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"51b2401531d8b6c49b8f830650aa724096e96f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 12:24:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa716dbfd3f74dd09a6f9b3c376d62fa : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d185f6282e58434793be16b502024da3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bff3a71d6ba4e4f909309177ed72f10 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8a96f1f8e63402eb2ea00fe5243eb60 : SUCCESS in 4m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fac51843a1b948289da4186254a48b42 : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf22be66ac654106b10579c4ab2b6efd : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03a6c6e4b9a642518b2f65adc9871b0d : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"e0012af3f704445f58433d1fee567aeafbfdbf45","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-03 13:15:42.000000000","message":"Removed Code-Review-2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"27ac012f526cdc0791eb50d41a214e424a951f17","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-03 13:15:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9fc8f50affc515c59f38cfb653d2aefda6626ad9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-22 11:13:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3aede7f8031c989345b81fd36e49d858d14e46ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:13:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3024f2795623a40099d176f8cefa7a3d6d9222dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:23:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fb39e69addf506dbb3543b623297176c532bf941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:23:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ab7c4b809d41d2a6732658ccd5c38c : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373e68a794b14cdb9528fa2e1ba89b63 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f11167c8a0c4fad9c6d78a7d3564f4c : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1d7737aeb1849d099591a2236628d2a : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5035ae09e69a40b496d5b78bffe3f559 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da369b9b5fbb41779cef4ba6c1d4387c : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbbff97e121d47f7a3a9190d700125c8 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"998e9805ba40c7153867398b129110a47871ef8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 11:26:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213f580234694a4fb0e7c03d28d537de : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4619d3ab51554f0fb15989eed9e83677 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85abe44233bee73148a895b21c4e965d310f3e70","revisions":{"7f8bf76a8792a431d4c7ff054e4feb05332d4df7":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:37:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/742147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/47/742147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/1"}}},"commit":{"parents":[{"commit":"d6b609727f39e31b8681427e079d268d911e20e8","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d6b609727f39e31b8681427e079d268d911e20e8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:37:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:37:18.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: I8d5290d62b12a7850fc932c3778cc37951dd67ed\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/7f8bf76a8792a431d4c7ff054e4feb05332d4df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/7f8bf76a8792a431d4c7ff054e4feb05332d4df7"}]},"branch":"refs/heads/master"},"22a742013f846c9369764d97cec486d432c360a9":{"kind":"REWORK","_number":2,"created":"2020-09-03 12:10:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/742147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/47/742147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/2"}}},"commit":{"parents":[{"commit":"d6b609727f39e31b8681427e079d268d911e20e8","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d6b609727f39e31b8681427e079d268d911e20e8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:37:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-03 12:10:02.000000000","tz":120},"subject":"dding pre-commit","message":"dding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I8d5290d62b12a7850fc932c3778cc37951dd67ed\nCo-authored-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/22a742013f846c9369764d97cec486d432c360a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/22a742013f846c9369764d97cec486d432c360a9"}]},"branch":"refs/heads/master"},"85abe44233bee73148a895b21c4e965d310f3e70":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-03 12:11:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/742147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/47/742147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/742147/3"}}},"commit":{"parents":[{"commit":"d6b609727f39e31b8681427e079d268d911e20e8","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d6b609727f39e31b8681427e079d268d911e20e8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:37:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-03 12:10:56.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I8d5290d62b12a7850fc932c3778cc37951dd67ed\nCo-authored-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/85abe44233bee73148a895b21c4e965d310f3e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/85abe44233bee73148a895b21c4e965d310f3e70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
