)]}'
{"id":"openstack%2Foslo.upgradecheck~763484","triplet_id":"openstack%2Foslo.upgradecheck~master~Ifa71c20879c964f794a40b981744d95db4ab1602","project":"openstack/oslo.upgradecheck","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"Ifa71c20879c964f794a40b981744d95db4ab1602","subject":"Add common checks for json to yaml policy file migration","status":"MERGED","created":"2020-11-20 00:32:37.000000000","updated":"2020-11-24 16:00:35.000000000","submitted":"2020-11-24 15:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763484-policy-json-to-yaml","meta_rev_id":"321b5e3a497138ccdb87577181cce5d7efdba62e","_number":763484,"virtual_id_number":763484,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-24 15:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-11-24 15:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-11-23 17:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-11-24 15:35:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 01:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 13:12:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-24 15:35:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"dfae092d3bf4a82c6dadccfa85e0708cfd712972","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-20 00:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f205450a42f83d1ed67fbcc1941a224d3b4a6205","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-20 00:32:55.000000000","message":"Topic set to policy-json-to-yaml","accounts_in_message":[],"_revision_number":1},{"id":"ee6e5235cdaa943351ff3c2dac1702dab0c41dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:41:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bebbc80f474c1abed9617a18974ec7 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1af416635a4c40ade054a6a07ffb7e : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df712ae0ff0d4fb1841996ccdde7d05c : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a9f688afdad422f9ecdfc1cd4e3beb2 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e66041123cf74f598d1c9609a7c5546f : SUCCESS in 4m 59s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd4b122c1ed24d89a3e330c3e6870787 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba007a037d04786afbf94a8961713d4 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7d2535be13f4c78afef8fb906eb29ca : FAILURE in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"003a95a73f5abb3922d5339295f191e63c3fae09","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-20 00:55:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e515f21e12e9a68e6930ef78665f4ef683b119d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:06:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3067f50bda040438f157b2ac5b18816 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a3c14af185442d95e47ba8e7d997e3 : SUCCESS in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3726414b4c35426584c919f723be1f80 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f7130a1182048aabf9e3bc0f66afa38 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ce9224aef03435d90ceb48cb55f89df : SUCCESS in 3m 35s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/891cfa1aa43148488e1a3b0e5b9c8bc1 : SUCCESS in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f23ee663432a46fca6648580f8fcbaec : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f08f36de5ea47be9f34a896bf28020a : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"8695abf113b45643a80e5a5d2a2b7cbf6f9fe2ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-20 13:12:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nLGTM, just few minor typos (-1 related)","accounts_in_message":[],"_revision_number":2},{"id":"672369a007fc71544f336c32163b863a097cae26","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-21 19:12:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c19a3879a4f3bccbafb35779409b028fb39a63ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-21 19:12:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b4edc68aaaff8f659565a66f5667ad5515320286","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-22 23:37:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"16ecff9b8a11074f37ea98820e5879473aca3aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-22 23:46:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd369f819b34680875a5e3c98b1135f : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8ed7985ef249e7a52e0aa49ef3329c : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c393b2b87c0b466e963de4a8efbe2cee : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c201c06829b24c52a20ced6681b89f8b : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7545d5dc39874a0db8399acea76f86ab : SUCCESS in 2m 37s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/d39e4f948786402382f821c6dd2ad5b5 : SUCCESS in 1m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24b235ad6b72477e84df1dad385c37ad : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb6369e6144471a89d788e8e3081544 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e836b854ebac6e27104b6b388755955c97ea8473","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-23 17:54:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"34b98b004091f468573d6fa324cf96cdfb23da09","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-24 15:35:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2b806c89ac506d64e409d060ecefe9ea36d15698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:37:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2acecb7e65411d1a96fa262fb95a39499aaea42b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:52:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37328381d7944e4fa795027d03de8256 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32c6185c4554b8ca8b7177a610d7c69 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b5bab7c9e1f488796bdf762d700757b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8717b5bcd6ff41edb62d16e2b239a884 : SUCCESS in 3m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5a26df2f20640ddaa54e5b8d2af57e4 : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80335b5a3c0d49b4abf715e824561440 : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a74cf5e65af41519d571e3ac1fc74d3 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4fec6518209f4029c134d5587e7e9bd3de64d402","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:53:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7032c0b10b946cd588d017284887a40615b65fa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:59:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fd9dca1ea74900a8443c87b2897fa8 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3525dcdf3fc247d09ab70d51584b3728 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3},{"id":"7018f7e7f4564e621c56c41ec9a19f92ae3c548a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:59:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fd9dca1ea74900a8443c87b2897fa8 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3525dcdf3fc247d09ab70d51584b3728 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3},{"id":"321b5e3a497138ccdb87577181cce5d7efdba62e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:00:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fd9dca1ea74900a8443c87b2897fa8 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3525dcdf3fc247d09ab70d51584b3728 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c70506727800d99fbbe8133f0bf0ccb1e70d690a","revisions":{"9a02d42dd4df6bbe7dff7768d8f0037090ae0d46":{"kind":"REWORK","_number":1,"created":"2020-11-20 00:32:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/763484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/84/763484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/1"}}},"commit":{"parents":[{"commit":"ff5ad7b821c58ed552171c67ecee23d5b3e9fe95","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/ff5ad7b821c58ed552171c67ecee23d5b3e9fe95"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-20 00:27:32.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-20 00:32:36.000000000","tz":-360},"subject":"Add common checks for json to yaml policy file migration","message":"Add common checks for json to yaml policy file migration\n\nAs part of community goal[1] for migrating the policy file\nfrom JSON format to YAML we need to add upgrade check to\ndetect any JSON formatted file. That check needs to be\ndone in each service side which are deprecating the JSON\nformat. So to avoid uplicate code, this commit add that\ncheck in oslo_upgradechecks and that can be enabled from\nservice side during they migrate to YAML format.\n\nService will enable it like below:\n\n    _upgrade_checks \u003d (\n        (_(\u0027Policy File JSON to YAML Migration\u0027),\n            (common_checks.check_policy_json, {\u0027conf\u0027: CONF})),\n    )\n\nAlso modify the check() method to accept the **kwargs for checks.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ifa71c20879c964f794a40b981744d95db4ab1602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9a02d42dd4df6bbe7dff7768d8f0037090ae0d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9a02d42dd4df6bbe7dff7768d8f0037090ae0d46"}]},"branch":"refs/heads/master"},"94e77fe2cc619da709d2375a7358780ac1c78c3b":{"kind":"REWORK","_number":2,"created":"2020-11-20 00:55:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/763484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/84/763484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/2"}}},"commit":{"parents":[{"commit":"ff5ad7b821c58ed552171c67ecee23d5b3e9fe95","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/ff5ad7b821c58ed552171c67ecee23d5b3e9fe95"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-20 00:27:32.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-20 00:55:03.000000000","tz":0},"subject":"Add common checks for json to yaml policy file migration","message":"Add common checks for json to yaml policy file migration\n\nAs part of community goal[1] for migrating the policy file\nfrom JSON format to YAML we need to add upgrade check to\ndetect any JSON formatted file. That check needs to be\ndone in each service side which are deprecating the JSON\nformat. So to avoid uplicate code, this commit add that\ncheck in oslo_upgradechecks and that can be enabled from\nservice side during they migrate to YAML format.\n\nService will enable it like below:\n\n    _upgrade_checks \u003d (\n        (_(\u0027Policy File JSON to YAML Migration\u0027),\n            (common_checks.check_policy_json, {\u0027conf\u0027: CONF})),\n    )\n\nAlso modify the check() method to accept the **kwargs for checks.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ifa71c20879c964f794a40b981744d95db4ab1602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/94e77fe2cc619da709d2375a7358780ac1c78c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/94e77fe2cc619da709d2375a7358780ac1c78c3b"}]},"branch":"refs/heads/master"},"c70506727800d99fbbe8133f0bf0ccb1e70d690a":{"kind":"REWORK","_number":3,"created":"2020-11-21 19:12:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/763484/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/84/763484/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/763484/3"}}},"commit":{"parents":[{"commit":"ff5ad7b821c58ed552171c67ecee23d5b3e9fe95","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/ff5ad7b821c58ed552171c67ecee23d5b3e9fe95"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-11-20 00:27:32.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-11-21 19:11:55.000000000","tz":0},"subject":"Add common checks for json to yaml policy file migration","message":"Add common checks for json to yaml policy file migration\n\nAs part of community goal[1] for migrating the policy file\nfrom JSON format to YAML we need to add upgrade check to\ndetect any JSON formatted file. That check needs to be\ndone in each service side which are deprecating the JSON\nformat. So to avoid duplicate code, this commit add that\ncheck in oslo_upgradechecks and that can be enabled from\nservice side during they migrate to YAML format.\n\nService will enable it like below:\n\n    _upgrade_checks \u003d (\n        (_(\u0027Policy File JSON to YAML Migration\u0027),\n            (common_checks.check_policy_json, {\u0027conf\u0027: CONF})),\n    )\n\nAlso modify the check() method to accept the **kwargs for checks.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ifa71c20879c964f794a40b981744d95db4ab1602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c70506727800d99fbbe8133f0bf0ccb1e70d690a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c70506727800d99fbbe8133f0bf0ccb1e70d690a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
