)]}'
{"id":"openstack%2Foslo.upgradecheck~775133","triplet_id":"openstack%2Foslo.upgradecheck~master~Id484bf8fe017bf8f6784155462640840b2053677","project":"openstack/oslo.upgradecheck","branch":"master","topic":"prettytable-2","hashtags":[],"change_id":"Id484bf8fe017bf8f6784155462640840b2053677","subject":"Uncap PrettyTable","status":"MERGED","created":"2021-02-11 12:03:36.000000000","updated":"2021-04-16 11:03:47.000000000","submitted":"2021-04-16 11:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775133-prettytable-2","meta_rev_id":"1f9d2f79a6b07d273f82acd80f13cb519dde7644","_number":775133,"virtual_id_number":775133,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 11:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-11 12:47:04.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-02-18 02:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-04-16 10:43:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2021-02-18 02:30:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 12:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 12:47:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-18 02:30:44.000000000","updated_by":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"reviewer":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"state":"REVIEWER"}],"messages":[{"id":"5ccd98ab752cb0404bcee5822554bbeefaf0fc8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-11 12:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7cb0f3bf4b22829b77ebe64f197e3e6a8ec33d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 12:15:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/893fa1d07dc7409291467567d03b2579 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4f6c241c06496ea4a0c33c7bf68053 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d145591e60db47369ff695940c044c8a : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a09aead1ee46388171c1c7a5f88b3e : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ea5bc25afb746beaf6ebdaf988d5988 : SUCCESS in 4m 13s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/7362d87cc9b64babb74a3e5a2d70f9c6 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e03cf9187a4f27b9f92dc0f50e750dda5dea75c5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 12:47:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9391bb4988dc90ae1e75c227e1e0cdc3ff10b72a","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-18 02:30:44.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"5e6f6a39af75536d1282023c15f9e074765658eb","author":{"_account_id":31882,"name":"Wander Way","email":"wanderwayout@gmail.com","username":"wanderway"},"date":"2021-02-18 02:30:54.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9e147fb2462f19b8265a31b23577a1a4ea45dfd7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:43:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58e960fabf957147787874773f820096c7ddc4db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:43:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9bde1858a4ddfa2fa8de6dad8dc9cf5fff303ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5859cf3bfaf64222a18d9e8601361507 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed92f519ba14e8d8f8912896fc846ae : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5409d1cfc24e2ca5d2af915b9ace1d : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5268033526294dc38cf3465e2412283b : SUCCESS in 3m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a7107bc44924137a1b0ffc981490e36 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"df8d00a47d44f71ba21cb00ce737b5228fec5193","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:02:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f9d2f79a6b07d273f82acd80f13cb519dde7644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a4045ce0c34307822efca152fc21e0 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f234316cce235d33b867a5c485b95b581e0f4702","revisions":{"f234316cce235d33b867a5c485b95b581e0f4702":{"kind":"REWORK","_number":1,"created":"2021-02-11 12:03:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/33/775133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/33/775133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/775133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/775133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/775133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/33/775133/1"}}},"commit":{"parents":[{"commit":"456bd8521f2a334eb056dacdc44320f36a64d223","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/456bd8521f2a334eb056dacdc44320f36a64d223"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-11 11:35:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-11 12:01:55.000000000","tz":0},"subject":"Uncap PrettyTable","message":"Uncap PrettyTable\n\nThis is now maintained as a Jazzband project [1].\n\n[1] https://github.com/jazzband/prettytable\n\nChange-Id: Id484bf8fe017bf8f6784155462640840b2053677\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/f234316cce235d33b867a5c485b95b581e0f4702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/f234316cce235d33b867a5c485b95b581e0f4702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
