)]}'
{"id":"openstack%2Foslo.upgradecheck~786615","triplet_id":"openstack%2Foslo.upgradecheck~stable%2Fwallaby~I3eafc92ba7c7305b1a73d3f96c427ed12c96dbf7","project":"openstack/oslo.upgradecheck","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I3eafc92ba7c7305b1a73d3f96c427ed12c96dbf7","subject":"Dropping lower constraints testing","status":"MERGED","created":"2021-04-16 10:07:41.000000000","updated":"2021-04-16 10:41:07.000000000","submitted":"2021-04-16 10:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786615-create-wallaby","meta_rev_id":"436f685f3c9e0c7a6ca8cf12292351eee2bfadca","_number":786615,"virtual_id_number":786615,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 10:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-16 10:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-16 10:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 10:07:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-16 10:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfcbd353b6372f8d9606a63d5d89f37ec995d624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3357dc9167bd9f93eac696ef2bc8f1e8e75736c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:08:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8ada0f4ee67e36473b18f3decfdbb915576b476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ff51d130c64b54aa69a73030e3978b : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b221580e8f34284960af44692dbce6e : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4490693f765b41448bf28c76dd4aade6 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/431176e1dd9640348231598c2577a11e : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd582d2bec64daab2d44849a992307e : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b866503b1bb5ade034fc25e49a461a243f01271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77bff4de922e6fc530821992b077b0c058b584f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:34:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4978474ec1344da9eebe9167bc19e76 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e92b5e5c3d74f389593870534cc05a9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/530a6a8223284b3ca2faa0305052dfe9 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de4e68892cf74fcfb706b73b7c2ce832 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"16bcb371be74cd8c6d2c4471f0be9f7172ae6d7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"436f685f3c9e0c7a6ca8cf12292351eee2bfadca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:41:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee994131ecdc4ae6b95408c576094b9d : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92e8272777c662ab80559bae34341d2058341789","revisions":{"92e8272777c662ab80559bae34341d2058341789":{"kind":"REWORK","_number":1,"created":"2021-04-16 10:07:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/786615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/15/786615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/15/786615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/15/786615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/15/786615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/15/786615/1"}}},"commit":{"parents":[{"commit":"9f95a6e19bbf17369cf9917480a9fe8b07f245ec","subject":"Register oslo policy opts explicitly for check_policy_json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/9f95a6e19bbf17369cf9917480a9fe8b07f245ec"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:13:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-16 10:06:36.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nNOTE(stephenfin): This change also includes change\nI996bcc6952a80aaccf50bb8ade579ebd960f4292 (\"Move flake8 as a pre-commit\nlocal target.\") which is also necessary to get requirements tests\npassing.\n\nChange-Id: I3eafc92ba7c7305b1a73d3f96c427ed12c96dbf7\n(cherry picked from commit a962943bdaa746ad238f003f56556eb02d1ea8e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/92e8272777c662ab80559bae34341d2058341789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/92e8272777c662ab80559bae34341d2058341789"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
