)]}'
{"id":"openstack%2Foslo.upgradecheck~856669","triplet_id":"openstack%2Foslo.upgradecheck~stable%2Fzed~Ie7f9f511bc442ad1b218c510fba4e4f74e2b474e","project":"openstack/oslo.upgradecheck","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"Ie7f9f511bc442ad1b218c510fba4e4f74e2b474e","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","status":"MERGED","created":"2022-09-09 09:18:35.000000000","updated":"2022-09-09 11:18:29.000000000","submitted":"2022-09-09 11:17:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856669-create-zed","meta_rev_id":"1c21cfeddb142797e3c74f06e0dcfb0aa6d5c259","_number":856669,"virtual_id_number":856669,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 11:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-09-09 10:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-09-09 10:58:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 09:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 10:58:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"14431872c1c9df5c07a0445662909f0659046630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-09 09:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87dc57b1925a501ad9e9ad9a176c976f6800b57f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:49:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7cb07fad07e4e359ce30ef2d0750630\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5661534ca49b4994a9a3925d94226b5f : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ab4a09daa148288d13cbc6327bfbbe : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f0fe370f0624f1baa3f4e33c6b09d14 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40830b93c7484ef693d9e4d567d4c867 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02591523bbe44c79af3810620335f7c1 : SUCCESS in 4m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37600cfc27184b4faaaaf6872bca18db : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e25c47b08f4f643e785e87d535ce19806d76611b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-09 10:58:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17f446ff05fb2b8e5f5c51f68178cff88aa3d0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:03:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6bfce5df0ed656af1df64931d8360c12540bebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:17:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7f328d969e45ab92cde0fc06c65740\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b23fc4d2c764644b591dc0139bff112 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c392e431584751bd9e5de96a0d05d5 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc3476128d144ad694fe86a0ce70c2b2 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65ef30b0a450402ea3a7ba1ae0b4d478 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f2dab45ea1b44f4b0efb060d24a92e3 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"49201ff079ac2a7e566908a40e7fb484ed715296","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:17:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c21cfeddb142797e3c74f06e0dcfb0aa6d5c259","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 11:18:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da07bc51e7345c6b4858764604892f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d53aba2e87a4328955c4e4126173915 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e92963d7c54a50bbdfde032cd40d5c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52526d77e0009f528fc4ee9d2b6784f2d726ce67","revisions":{"52526d77e0009f528fc4ee9d2b6784f2d726ce67":{"kind":"REWORK","_number":1,"created":"2022-09-09 09:18:35.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/856669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/69/856669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/856669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/856669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/856669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/69/856669/1"}}},"commit":{"parents":[{"commit":"6bb018ef2b254aab21a7d47a99dfe36dd73fd506","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/6bb018ef2b254aab21a7d47a99dfe36dd73fd506"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 09:18:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-09 09:18:34.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","message":"Update TOX_CONSTRAINTS_FILE for stable/zed\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/zed branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ie7f9f511bc442ad1b218c510fba4e4f74e2b474e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/52526d77e0009f528fc4ee9d2b6784f2d726ce67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/52526d77e0009f528fc4ee9d2b6784f2d726ce67"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
