)]}'
{"id":"openstack%2Foslo.upgradecheck~869347","triplet_id":"openstack%2Foslo.upgradecheck~master~Ibd7747f8b337c27f298c7e92ece41bb911cdf679","project":"openstack/oslo.upgradecheck","branch":"master","topic":"oslo-tox-4","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-27 12:04:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibd7747f8b337c27f298c7e92ece41bb911cdf679","subject":"Get ready for tox 4","status":"ABANDONED","created":"2023-01-05 10:45:40.000000000","updated":"2023-02-27 12:04:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4666dcf9b1befd057e36e7c01dc93d23c41ce032","_number":869347,"virtual_id_number":869347,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-10 11:10:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 11:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb00264e8e42603f49058bf6a98cf7a48378d156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-05 10:45:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f41c585e2839a5a57d17673f25a07279479cc0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 11:03:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e770dd4d909470898c38a76db20dcd4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28b92666a578481e94456685398792e9 : SUCCESS in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033aee4b286341e482b831ca901ebff0 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b3512e5d704a60ad595f6704b5d22f : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc082b869e6e4cc1af3681e72ac8d9b5 : FAILURE in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72c57f533a9d41ae862c8f17865ab6f3 : FAILURE in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8a921508b8f4b45b4fee6d52edff210 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ac426ed118447c33f85d3eddceffa904e56735fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-10 11:00:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e92a8b155e9df30b84fd4479b0f30b63335715a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 11:10:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59d9f7e2063e45cd96a981ae2f0cef07\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d754307201a14f48ab185441c3113ecc : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68723be906174e4ea7025a7e34c9f5a2 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9100032344b8400baecc072f69a2a626 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e933d5ef1d8c423a82ff71d75c090637 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b705d91a5de74afc8e8b16346441b7d1 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1a0a464e9ee4b1489845367d8eb2300 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4666dcf9b1befd057e36e7c01dc93d23c41ce032","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 12:04:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98d264cb3729203aa2c6478b2e7c67be262123e0","revisions":{"2ef18b72e6c4d4113fe11c558581b608991b7b83":{"kind":"REWORK","_number":1,"created":"2023-01-05 10:45:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/869347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/47/869347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/1"}}},"commit":{"parents":[{"commit":"cd42c3029929040ba8423ac30c0d0003ce44ebe1","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cd42c3029929040ba8423ac30c0d0003ce44ebe1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-05 10:43:05.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-05 10:45:39.000000000","tz":60},"subject":"Get ready for tox 4","message":"Get ready for tox 4\n\nChanges:\n- made 4.0.0 the tox minversion\n- disabling auto discovery by explicitly setting py_modules\u003d[] in\n  setup.py to prevent a setuptools issue\n  https://github.com/pypa/setuptools/issues/3197#issuecomment-1078770109\n\nThis patch makes tox 4 the default so that we can hopefully catch\nproblems locally before they block the gate.\n\nChange-Id: Ibd7747f8b337c27f298c7e92ece41bb911cdf679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/2ef18b72e6c4d4113fe11c558581b608991b7b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/2ef18b72e6c4d4113fe11c558581b608991b7b83"}]},"branch":"refs/heads/master"},"98d264cb3729203aa2c6478b2e7c67be262123e0":{"kind":"REWORK","_number":2,"created":"2023-01-10 11:00:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/47/869347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/47/869347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/47/869347/2"}}},"commit":{"parents":[{"commit":"cd42c3029929040ba8423ac30c0d0003ce44ebe1","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cd42c3029929040ba8423ac30c0d0003ce44ebe1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-05 10:43:05.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-10 10:58:32.000000000","tz":60},"subject":"Get ready for tox 4","message":"Get ready for tox 4\n\nChanges:\n- made 4.0.0 the tox minversion\n- disabling auto discovery by explicitly setting py_modules\u003d[] in\n  setup.py to prevent a setuptools issue\n  https://github.com/pypa/setuptools/issues/3197#issuecomment-1078770109\n- disable skipsdist to avoid failure with missing package\n\nThis patch makes tox 4 the default so that we can hopefully catch\nproblems locally before they block the gate.\n\nChange-Id: Ibd7747f8b337c27f298c7e92ece41bb911cdf679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/98d264cb3729203aa2c6478b2e7c67be262123e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/98d264cb3729203aa2c6478b2e7c67be262123e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
