)]}'
{"id":"openstack%2Foslo.upgradecheck~888717","triplet_id":"openstack%2Foslo.upgradecheck~stable%2Fvictoria~Iebac095f3b14f65bba868bf30807742d84ab5ef9","project":"openstack/oslo.upgradecheck","branch":"stable/victoria","topic":"gate-fix-cap-virtualenv","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-02-20 09:44:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iebac095f3b14f65bba868bf30807742d84ab5ef9","subject":"[stable-only] Cap virtualenv/setuptools","status":"MERGED","created":"2023-07-17 15:30:06.000000000","updated":"2024-02-20 09:46:52.000000000","submitted":"2024-02-20 09:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888717-gate-fix-cap-virtualenv","meta_rev_id":"cd2df9b6addfc97a106fc06be189435ce8de1aee","_number":888717,"virtual_id_number":888717,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-20 09:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-20 09:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-20 09:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 15:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 09:34:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"e34d1d84d71858447cc7c7c55f59e323dc8ce266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-17 15:30:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a0dc0bd9f55c944c432e03cca259f59af6abbaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 15:39:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c470c9849a4b413f9f740f65bcd8b501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4655ec409143eda91961d39d835e8e : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b784aa2fa6ed42d080766c1952c5d2d5 : SUCCESS in 2m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beb27176b31d4a9e82ffa59ab9c3d9ba : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f14f7102d504e359ee687ff9ee1708d : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f9d223dbedb462597f12dfea211ab36 : SUCCESS in 2m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1177c1ca0b074b28a21b951d67c8175d : FAILURE in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0780da552bccec9ecef8c211b4e9fb36fb7c06e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-17 15:47:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"66df065824c45549b770afef2ab7d70ba3d87046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 16:02:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce504508c8bd466eb52e0fe3fbcbbc50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc77f0e01b5f4e1b82770b1a654dce5a : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c141d0ce4c495f95372d3396174ef0 : SUCCESS in 2m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cbb5b6fbf1046c3bdbb6be4899f0998 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60b1c867312a4044a68aeab3f68c9409 : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63944f43f3dc40a7b784c2a4df8c19d7 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5561a57d90d3860b5927f424e66e5406a533c8f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 09:34:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99ee07ece9424b85b5beaa5ddf42670ea915b78b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 09:34:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"012e9329d86ef22c858c96bb907e7397d757a0f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 09:44:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ce69a2cf5a4f089d9fcd47967a9656\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50f91a8890043cabee55d26e047ec1d : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65758d398d6b4e979df98ae7ee853d14 : SUCCESS in 2m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a98a3da352c4da69a9ec6684f124f96 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b16a4a292394cf3a807d336f9654a7f : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2f0be2a3814044bbdb79250b990e50 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7dfa7c6f2b7274d43dfaefa9deae2a0398c789a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 09:44:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd2df9b6addfc97a106fc06be189435ce8de1aee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 09:46:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce12e70824740d89a0f93183de46bc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de55f3d15a7f4bc8b3bdfbd645f38e5f : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aa29cfcd8b54193b953de7a7ea08d3e : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"912d6544ada2f25f8d2d8e94d240050f06a805e4","revisions":{"cb65641087eba83fec00fc11ec1b7da522717e69":{"kind":"REWORK","_number":1,"created":"2023-07-17 15:30:06.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/888717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/17/888717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/1"}}},"commit":{"parents":[{"commit":"29acd2fd980c837a7403bc7f7cea21318c098cda","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/29acd2fd980c837a7403bc7f7cea21318c098cda"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:29:32.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:29:51.000000000","tz":120},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: Iebac095f3b14f65bba868bf30807742d84ab5ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cb65641087eba83fec00fc11ec1b7da522717e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/cb65641087eba83fec00fc11ec1b7da522717e69"}]},"branch":"refs/heads/stable/victoria"},"912d6544ada2f25f8d2d8e94d240050f06a805e4":{"kind":"REWORK","_number":2,"created":"2023-07-17 15:47:55.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/17/888717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/17/888717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/17/888717/2"}}},"commit":{"parents":[{"commit":"29acd2fd980c837a7403bc7f7cea21318c098cda","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/29acd2fd980c837a7403bc7f7cea21318c098cda"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:29:32.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:46:04.000000000","tz":120},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\nAlso dropping lower constraints testing as it was dropped on master and\nnewer stable branches already.\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: Iebac095f3b14f65bba868bf30807742d84ab5ef9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/912d6544ada2f25f8d2d8e94d240050f06a805e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/912d6544ada2f25f8d2d8e94d240050f06a805e4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
