)]}'
{"id":"openstack%2Foslo.upgradecheck~969097","triplet_id":"openstack%2Foslo.upgradecheck~master~Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947","project":"openstack/oslo.upgradecheck","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-05 16:30:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-12-05 16:47:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947","subject":"Add typing","status":"MERGED","created":"2025-12-01 14:46:57.000000000","updated":"2025-12-05 16:48:32.000000000","submitted":"2025-12-05 16:47:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969097-typing","meta_rev_id":"85cd76f7fcd431342670b973b58bd76261c961fd","_number":969097,"virtual_id_number":969097,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 16:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-05 16:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-05 16:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 14:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 16:09:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"886b9147c94d17ab347673a651a45d53d561e1d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 14:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c3f78dc41add5516e978ce1d477cb3a17bb5c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 14:55:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f402de2f36014f888d52f941c24444e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3defebe7448b44588e263abed2a287c5 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128985219997450aaf04c7a174410d69 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db177baeb32c4e80813b27921be71162 : FAILURE in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91e25336eec64525b87e15e83df39f89 : FAILURE in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f006d667e36428682b1ab6b520bebde : FAILURE in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/241a3ff814b2423296479e77f542033d : SUCCESS in 1m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d760bb464b4b4877836394d2de384849 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1},{"id":"73a913f6b3d88cff7953275e62f6c8da8ce952dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 16:43:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a12773c253e023b2103ad3656eac6d703c20e665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:54:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be276ddfe85e4519a2cd00c5236d8408\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9fc67fc6efec4e15ac4e0a4f60e5c986 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f3dc65e2ab4be9a981aed314dbaa3a : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182e12f4dd854c81a54f4402537be132 : FAILURE in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbceae95c5814a2883843348b384a9ad : FAILURE in 1m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c40969baa9524a808b7af201a2f8523f : FAILURE in 5m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa365228619249018c35a3d8288e8bc6 : SUCCESS in 1m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74972bb02b37433292e3be9c83738222 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"cab683e4dfda9842710fbe707f9c50e18b9acd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 17:29:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"630faf85240afe8e6fd9d93d23408adda145d9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 17:52:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f46227b93a246f88134300a4654b709\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1538032a0bea4a19a6196ae090c68dd4 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cf784483574818b8274e6fb7020148 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842d94ce501743d5b7d20bf024176ce6 : FAILURE in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/071b322ee8d141a9b5fa65f5c99b3ba7 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2ef720746aa49218d38c5a4d3e7eec4 : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fac026bede9b481a88082e4de2839c9c : SUCCESS in 2m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f502c24f271b4b158280bbd2a0fde714 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"97d35062f4068ae3ce94b151f5c58a9bba2b22e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 14:33:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c0b7b571ca2b683109748a50f9414feab09210ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:43:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442d495c898648b494cd7eda8add1b75\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0d8d40811584357885500dbf03d26fe : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a227f959a75641b1a088d655c5fa5b4f : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a593e8ae3d40828098e5b1b1b7d5f2 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccca3189b34c4cc9b5322e422aa59eba : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e906b8cf1dd4bd2ab8dc46a22005ec2 : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16ed611c8701409fbea0a93b11a88a86 : SUCCESS in 1m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a5e39057bd84428a27bb4af65c343e7 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":4},{"id":"cc1c41ab6e43fbcd8d06f829336cc8694de9e916","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-12-04 16:09:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"933201109e374507db6d5bd7bc12a4f3db5c5bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:19:44.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bea617fa1b875417f3899865733f4d29327eccef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:46:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f904f31972e4585a35c72030129a924\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82a21e1ebd7e4a298522603eb64763dc : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d44935fff3472a8290a50a80e9a072 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1e879814be40f1b90f5871b7f002da : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64597a7bd336449cba215d758963a990 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c96762e682694366a3293d10137e0a53 : SUCCESS in 5m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d17abde7907a4d3c8750218e222e1794 : SUCCESS in 2m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d827d35b74714ea0b533c03c1f579861 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":5},{"id":"333b6b440f36d2b5f6ce63ee8ea1d25113b5faf4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-05 16:30:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"26121973087520e42964873aca8bec9a198e1ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:31:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a80a8e02cbd6660ceb5cc3c07672f09d79233a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:47:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e811e13eb11411eb8abf9dd54a42a71\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f666a04220c4986bb814bd0988d0b79 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22fdb926c494433baeda918ca709392 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37cda6589b342b991bbeb5dcc4179c4 : SUCCESS in 1m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db7013db83dc4b2aadd7255f409ed254 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7d71232daad4b92891734f9565999b8 : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/210f8822080d4bb38231f1828b0c0453 : SUCCESS in 2m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7ac58368acf4acda019f6221d273ba4 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":5},{"id":"74c963db5146ba671a46d38b9a8f96a49f5e6ccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:47:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"85cd76f7fcd431342670b973b58bd76261c961fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:48:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab0d6875ca404cf4836f53a950f4e19c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd8cd54785d443b9c6780aff9108f19 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/987bf54d9d024fdf8e45dd5f6805ae2b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c218daced5569d0c7c6b1f613a8a9f6ad0f8a49f","revisions":{"678f3ff9d07f2b457b5e5b0d44f5902650dfc384":{"kind":"REWORK","_number":1,"created":"2025-12-01 14:46:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/969097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/97/969097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/1"}}},"commit":{"parents":[{"commit":"d8dea5bafccfa4e4f5c87e204a7a59f11e704500","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d8dea5bafccfa4e4f5c87e204a7a59f11e704500"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:19:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:45:54.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nChange-Id: Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/678f3ff9d07f2b457b5e5b0d44f5902650dfc384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/678f3ff9d07f2b457b5e5b0d44f5902650dfc384"}]},"branch":"refs/heads/master"},"c250e7154c4a5e43f5a29a7483435acd7ce40b08":{"kind":"REWORK","_number":2,"created":"2025-12-01 16:43:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/969097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/97/969097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/2"}}},"commit":{"parents":[{"commit":"d8dea5bafccfa4e4f5c87e204a7a59f11e704500","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d8dea5bafccfa4e4f5c87e204a7a59f11e704500"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:19:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 16:43:48.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nChange-Id: Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c250e7154c4a5e43f5a29a7483435acd7ce40b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c250e7154c4a5e43f5a29a7483435acd7ce40b08"}]},"branch":"refs/heads/master"},"d42040538561678a2e98819e5d804c4c6b0aebbe":{"kind":"REWORK","_number":3,"created":"2025-12-01 17:29:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/969097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/97/969097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/3"}}},"commit":{"parents":[{"commit":"d8dea5bafccfa4e4f5c87e204a7a59f11e704500","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d8dea5bafccfa4e4f5c87e204a7a59f11e704500"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:19:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 17:28:55.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nChange-Id: Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d42040538561678a2e98819e5d804c4c6b0aebbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d42040538561678a2e98819e5d804c4c6b0aebbe"}]},"branch":"refs/heads/master"},"950007e86bdfc67f6a7725a60ea69f6b16e18e95":{"kind":"REWORK","_number":4,"created":"2025-12-04 14:33:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/969097/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/97/969097/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/4"}}},"commit":{"parents":[{"commit":"d8dea5bafccfa4e4f5c87e204a7a59f11e704500","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d8dea5bafccfa4e4f5c87e204a7a59f11e704500"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:19:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-04 14:33:03.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nChange-Id: Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/950007e86bdfc67f6a7725a60ea69f6b16e18e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/950007e86bdfc67f6a7725a60ea69f6b16e18e95"}]},"branch":"refs/heads/master"},"c218daced5569d0c7c6b1f613a8a9f6ad0f8a49f":{"kind":"REWORK","_number":5,"created":"2025-12-04 18:19:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/969097/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/97/969097/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/97/969097/5"}}},"commit":{"parents":[{"commit":"d8dea5bafccfa4e4f5c87e204a7a59f11e704500","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/d8dea5bafccfa4e4f5c87e204a7a59f11e704500"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-01 14:19:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-12-04 18:19:35.000000000","tz":0},"subject":"Add typing","message":"Add typing\n\nChange-Id: Iee0e8436ac40054a2b7f2d0019efe0bc72d6f947\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c218daced5569d0c7c6b1f613a8a9f6ad0f8a49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c218daced5569d0c7c6b1f613a8a9f6ad0f8a49f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
