)]}'
{"id":"openstack%2Foslo.upgradecheck~979284","triplet_id":"openstack%2Foslo.upgradecheck~master~Ia06efae0cf5bca1dc136265ab3a65aa2bdd62b49","project":"openstack/oslo.upgradecheck","branch":"master","hashtags":[],"change_id":"Ia06efae0cf5bca1dc136265ab3a65aa2bdd62b49","subject":"mypy: Ignore missing imports","status":"ABANDONED","created":"2026-03-06 15:24:18.000000000","updated":"2026-03-06 15:41:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d5cc5cfe51fca361966a5463d5071bebad5aa84","_number":979284,"virtual_id_number":979284,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-06 15:32:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-06 15:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee69802add77b93935d0a6a7dc694083b3e0a296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-06 15:24:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02dd44bcafb4e44dafd61a787b0486c67dc2a6ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 15:32:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945b67e83c8749cba211ac7e183e555e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b359d154a57c4ae3b1ceec4e3ce5b3b4 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10dfe2ed117a46e1b5aa0a87c7eb7778 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5b4a76b05549bd8322c62106b7640e : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c49a870649f4b40b8177109ddbd5dfb : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e777f6e0714e4c6989bee194699e9d24 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/141ead1ad5ca42b4a1da65f5a38f24b9 : SUCCESS in 5m 47s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/07d5e5289a1445da8dba77faf0972499 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0d5cc5cfe51fca361966a5463d5071bebad5aa84","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-06 15:41:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"440d3b0a28e575f0d2cea1c012dabd73447857e4","revisions":{"440d3b0a28e575f0d2cea1c012dabd73447857e4":{"kind":"REWORK","_number":1,"created":"2026-03-06 15:24:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/979284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/84/979284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/979284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/979284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/979284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/84/979284/1"}}},"commit":{"parents":[{"commit":"b21f8e2cccd5642e9c2f707de91467826e41ec50","subject":"Add typing classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/b21f8e2cccd5642e9c2f707de91467826e41ec50"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-06 15:23:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-06 15:24:08.000000000","tz":540},"subject":"mypy: Ignore missing imports","message":"mypy: Ignore missing imports\n\n... to avoid immediate typing error caused by a new release with\ntyping support.\n\nChange-Id: Ia06efae0cf5bca1dc136265ab3a65aa2bdd62b49\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/440d3b0a28e575f0d2cea1c012dabd73447857e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/440d3b0a28e575f0d2cea1c012dabd73447857e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
