)]}'
{"id":"openstack%2Foslo.upgradecheck~994281","triplet_id":"openstack%2Foslo.upgradecheck~master~I3a34df059ba84c409b2949a037fb905ee88c3fbb","project":"openstack/oslo.upgradecheck","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-24 14:58:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a34df059ba84c409b2949a037fb905ee88c3fbb","subject":"Do not install code to build release notes","status":"MERGED","created":"2026-06-22 12:33:47.000000000","updated":"2026-06-24 15:00:50.000000000","submitted":"2026-06-24 14:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994281","meta_rev_id":"c2257df81be299fbb4d9dccca91e5b90640c3990","_number":994281,"virtual_id_number":994281,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 14:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-24 14:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-24 14:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 12:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 14:47:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f123e27af230cec31d61330aa0ac64734cd3d25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-22 12:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91937c1d4752c90cc4ad0ce64cbb8c34148d0d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:49:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e223f7dab6d4703ad02d02473ad40c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342d0c5b40d7476ab251f4cb18ff9677 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f119d78e5804473ab5713984abcbb583 : FAILURE in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36efa31acda540bc929d77089b9ac6d9 : SUCCESS in 1m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f4d7d9908af4668b956fc93e2a307b5 : SUCCESS in 2m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2701d17ce9ba447894af52419f074558 : SUCCESS in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84926cea305d446d81525e7c5c4828c2 : SUCCESS in 2m 58s\n\nWarning:\n  Comments left for invalid file oslo_upgradecheck/common_checks.py\n  Comments left for invalid file oslo_upgradecheck/tests/test_common_checks.py\n  Comments left for invalid file oslo_upgradecheck/tests/test_upgradecheck.py\n  Comments left for invalid file oslo_upgradecheck/upgradecheck.py","accounts_in_message":[],"_revision_number":1},{"id":"adb9c82c398cb24fb37e2e9d6fc8edfa2c30a459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-23 18:29:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4bf12953d8a8e61b34eee7c09964c1fad38e3d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 18:41:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca7bf1e0ca342bab7cd74bff3b488c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c584564e47f94c688578c10e72a2a9a0 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd70e1271754d818c42f7cbb93309dd : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b10bcd3d43b34176a8835c88c894bc01 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b47d4dbb6144a69b6a53547694d47fe : SUCCESS in 4m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cca372815ec3409989f9ca92b571a681 : SUCCESS in 5m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aaa379353d442a3881b1431f8cf3eed : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4f1837aa22b908f49efe42edc4cc6c84ed81053d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 14:47:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3592fbe764c27c6a5ef89d618d83b40f3e1df4e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:50:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a6fc2c393dd1b908ea20dc8b394ad0cce8697d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:58:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cee63e129e5f4ab79414b6475e8cc3ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc567c96a5ea4ab4893c43abeaa2f6b8 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b366914fc0746c7a99d110f1967812e : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ffb392332004904816dd54853477a89 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e30fd3da65494ee59fddc45b7ea5205a : SUCCESS in 2m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6746886f768040cea1bfb8551f291799 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8d4997f273505b7b909d5595c57a6d3bab35488f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:58:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c2257df81be299fbb4d9dccca91e5b90640c3990","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 15:00:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d5c3eaca9641abb21c7753bc8f5de4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11c8c79648745578bbcbe0f99310257 : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e18c1b0eef7247d9942331ed09ea8c8c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1ed846b90979ca24eaba4671bd8b80f2320abc0","revisions":{"3c890f14fd29dc0a2fa8f2ce97a839bd0f9000c2":{"kind":"REWORK","_number":1,"created":"2026-06-22 12:33:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/994281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/81/994281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/1"}}},"commit":{"parents":[{"commit":"8f39b19a5615e53d6ab6445d15e3764d8a06d607","subject":"Merge \"Update packaging configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/8f39b19a5615e53d6ab6445d15e3764d8a06d607"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:33:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:33:36.000000000","tz":540},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: I3a34df059ba84c409b2949a037fb905ee88c3fbb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/3c890f14fd29dc0a2fa8f2ce97a839bd0f9000c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/3c890f14fd29dc0a2fa8f2ce97a839bd0f9000c2"}]},"branch":"refs/heads/master"},"c1ed846b90979ca24eaba4671bd8b80f2320abc0":{"kind":"REWORK","_number":2,"created":"2026-06-23 18:29:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/994281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.upgradecheck","ref":"refs/changes/81/994281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.upgradecheck refs/changes/81/994281/2"}}},"commit":{"parents":[{"commit":"8f39b19a5615e53d6ab6445d15e3764d8a06d607","subject":"Merge \"Update packaging configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/8f39b19a5615e53d6ab6445d15e3764d8a06d607"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-22 12:33:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-23 18:29:30.000000000","tz":540},"subject":"Do not install code to build release notes","message":"Do not install code to build release notes\n\n... because we require only doc requirements.\n\nAlso remove explicit usage of requirements.txt because it is now\nautomatically detected and installed.\n\nChange-Id: I3a34df059ba84c409b2949a037fb905ee88c3fbb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c1ed846b90979ca24eaba4671bd8b80f2320abc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.upgradecheck/commit/c1ed846b90979ca24eaba4671bd8b80f2320abc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
