)]}'
{"id":"openstack%2Foslo.utils~1007393","triplet_id":"openstack%2Foslo.utils~master~I72ec1066a4c40d7b9bdafd00bb62756a17839dd9","project":"openstack/oslo.utils","branch":"master","topic":"py315","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-25 10:01:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I72ec1066a4c40d7b9bdafd00bb62756a17839dd9","subject":"Declare Python 3.15 support","status":"MERGED","created":"2026-09-25 04:48:43.000000000","updated":"2026-09-25 12:29:13.000000000","submitted":"2026-09-25 12:28:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007393-py315","meta_rev_id":"2e1dc6137213b27cfb2b69251ccb833dd0908821","_number":1007393,"virtual_id_number":1007393,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-25 12:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-25 08:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-25 08:13:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 06:34:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e34d6bcf037ac7bab4a89ed8ac07efe6c13a4531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 04:48:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76777bd647fc21dbc4d1af8467f97d67b5fccadf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:34:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b0e35d82c144c192afb00998a2fc9e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4bef093068874eb388af8902febc2efa : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d703f5e2b9394105b8ace487dc272006 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dc11ac8d1c648cd93fd0780354f5792 : SUCCESS in 1h 39m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbabff36afee450f842c35af0ba07d10 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614afb0384394929959bcd810cca957c : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4244cc98d7234e06843db20684674f54 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a34209fe9bf040dbbd002c7a2366acea : SUCCESS in 8m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0682e8a12609473c90809d81d08d0872 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ac115f3eb14b0daa072d9b83b7565d : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6a62e9e7dd24f45bcd998e995ee5ba0 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6fd7749f11e2f5e7dfcdc3d734425719543d3b29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 08:13:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3eb5646cf667f3389ff1592df11c5490b281cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:14:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4a940780f658b4a40ff7cdc9702e1413a31f0e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 09:54:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/693a59dbdee14edc960c3540cbe80df4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3299810361e04ea5a4d77a9eaa33cdf0 : SUCCESS in 1m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e8997a5cadf471f8ae6fef8c875c179 : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3a4a6b9a6b7456fbe965b90b93d47e9 : SUCCESS in 1h 35m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b75c1752cc427bbc040d7a4e3356f8 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8169120f44c94821aa2b5cc430382492 : FAILURE in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/858b28d804ca4b79aeb356777e0ef64f : SUCCESS in 8m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/87b61ea7dab14010ba7cdcca08514539 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cde93226de4fb29e6bc21f64e34c06 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff570fbb942d45e2860c9dcf7a637a24 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"97c0f87406c45770c79543d7f40e3f8148dfa33d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 10:01:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93607602f201694b804b2e3ee4f38006177176a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:36:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7140cd0d6302448682506533f3e98470\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da560c1c7d27468993674ce57dbf01a2 : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb29be0858284f7a9315fcced380f314 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80209a0210354e65a66631720b9c80ad : SUCCESS in 1h 29m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a9008eba2224046a7aed78405f6c720 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5764d10d8e6e4237a49810d54d427ca2 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19c61c1ccba5445d9e3c853b821ba3bc : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d161b37b2f0b4223a14f71f10c7fe14e : SUCCESS in 7m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ab1df27ab80245139142b1f3e873e3ca : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f6a1ede55348938443adaa5f8ba233 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da09f970eb4247ba8979b656d6c024fa : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a28af5a183b5dd7b3d04b157458128584de17405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 11:37:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"056389b8fc10262dd3f2e11ea06875fb006aa0a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:28:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33ff31f9e3a24426a7a5724a963a3d4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eca68ec6640a49b49a81e03cd8b7e90a : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ab05578a1fd4e499c9a7cd61319a95e : SUCCESS in 3m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63f9aed57e844258acc31456c2da1098 : SUCCESS in 49m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c82bfc748240c3b0d8be047f084ba3 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0212273740664ea98f536d590523b171 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36526d433e74487fbe75db2599fc1d56 : SUCCESS in 7m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ec447020ed314e128601140018b14049 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14890d96ccc14963ae35264661bd71bc : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/131e4144a53a42a7b77880b17254b1fa : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e82260df41785c3203d18186099fd7024822c9a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:28:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e1dc6137213b27cfb2b69251ccb833dd0908821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:29:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49d9d7fa1c5c45d9ab677a46678def4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06fe3071d1564c3b87fbb59d2b4533f5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dba67868cd143406d7b032d4ce039dfefd0ade31","revisions":{"dba67868cd143406d7b032d4ce039dfefd0ade31":{"kind":"REWORK","_number":1,"created":"2026-09-25 04:48:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/1007393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/93/1007393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/93/1007393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/93/1007393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/93/1007393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/93/1007393/1"}}},"commit":{"parents":[{"commit":"5f10aaacb05406a5ab690e8ba9deefe570e2b7df","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5f10aaacb05406a5ab690e8ba9deefe570e2b7df"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:48:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 04:48:33.000000000","tz":540},"subject":"Declare Python 3.15 support","message":"Declare Python 3.15 support\n\nIt\u0027s now voting in the -next job template.\n\nChange-Id: I72ec1066a4c40d7b9bdafd00bb62756a17839dd9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/dba67868cd143406d7b032d4ce039dfefd0ade31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/dba67868cd143406d7b032d4ce039dfefd0ade31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
