)]}'
{"id":"openstack%2Foslo.utils~119097","triplet_id":"openstack%2Foslo.utils~master~I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1","project":"openstack/oslo.utils","branch":"master","hashtags":[],"change_id":"I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1","subject":"Auto generate the API documentation","status":"ABANDONED","created":"2014-09-04 15:53:04.000000000","updated":"2014-11-07 10:23:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"814f89824e37ec449962d2ebe87103d94cb02e87","_number":119097,"virtual_id_number":119097,"owner":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2014-10-27 21:05:27.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-10-27 18:13:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-09-10 09:09:03.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2014-10-03 20:50:35.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2014-10-03 23:53:10.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2014-10-06 09:25:54.000000000","updated_by":{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},"reviewer":{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},"state":"REVIEWER"},{"updated":"2014-10-27 21:05:27.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"cc18388f3c32ca5c5ebb5e13b25c0bffbb74285c","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-09-04 15:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53e6e2225241dd832cc6a6c4874fc91cbe8132c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-04 22:19:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/1/check/gate-oslo.utils-pep8/b591eea : SUCCESS in 53s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/1/check/gate-oslo.utils-docs/a48e8c0/doc/build/html/ : SUCCESS in 32s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/1/check/gate-oslo.utils-python26/3288f5b : SUCCESS in 1m 30s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/1/check/gate-oslo.utils-python27/cacb10d : SUCCESS in 54s\n- check-tempest-dsvm-full http://logs.openstack.org/97/119097/1/check/check-tempest-dsvm-full/6c983e7 : SUCCESS in 51m 19s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/97/119097/1/check/check-tempest-dsvm-postgres-full/1ab2293 : SUCCESS in 1h 00m 16s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/97/119097/1/check/check-tempest-dsvm-neutron-full/0f36e2a : SUCCESS in 54m 07s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/97/119097/1/check/check-tempest-dsvm-neutron-heat-slow/19c79d9 : SUCCESS in 21m 02s\n- check-grenade-dsvm http://logs.openstack.org/97/119097/1/check/check-grenade-dsvm/acce3b0 : SUCCESS in 38m 06s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/1/check/gate-oslo.utils-python33/8917dd6 : FAILURE in 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce390caa7e0ab40981a21d4453ab06521b5f9477","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-09-05 01:34:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d2b457cd8063b1a21d246bd8545a6a9ba4de681","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-05 03:04:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-pep8/e440343 : SUCCESS in 1m 12s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/2/check/gate-oslo.utils-docs/e41b3b2/doc/build/html/ : SUCCESS in 1m 20s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python26/aa48e4d : SUCCESS in 1m 51s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python27/e5d7312 : SUCCESS in 1m 14s\n- check-tempest-dsvm-full http://logs.openstack.org/97/119097/2/check/check-tempest-dsvm-full/ed42dc6 : SUCCESS in 59m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/97/119097/2/check/check-tempest-dsvm-postgres-full/9029585 : SUCCESS in 1h 17m 45s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/97/119097/2/check/check-tempest-dsvm-neutron-full/eabf5ee : SUCCESS in 1h 05m 40s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/97/119097/2/check/check-tempest-dsvm-neutron-heat-slow/2c2a2c8 : SUCCESS in 24m 58s\n- check-grenade-dsvm http://logs.openstack.org/97/119097/2/check/check-grenade-dsvm/2d43986 : SUCCESS in 47m 24s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python33/cf68b3d : FAILURE in 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"361d6715feae2d521e6d2853621408fea1e1a2d2","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2014-09-10 09:09:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0932e39dfee1bb42c98905bed32a72a49f779fa1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-03 20:50:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"5ed9cdf7fcbfb3951df6ff722b23212ace62b9d7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-10-03 20:50:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5b34911144816d3289fbbbf3aab9eb14f68a5520","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-03 23:42:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.utils-docs http://logs.openstack.org/97/119097/2/gate/gate-oslo.utils-docs/7327719 : FAILURE in 46s\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/2/gate/gate-oslo.utils-pep8/e712de9 : SUCCESS in 42s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/2/gate/gate-oslo.utils-python26/31bf726 : SUCCESS in 59s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/2/gate/gate-oslo.utils-python27/2a60ca8 : SUCCESS in 1m 03s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/2/gate/gate-oslo.utils-python33/4782c7f : FAILURE in 19s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/2/gate/gate-tempest-dsvm-src-oslo.utils/2747ee5 : SUCCESS in 50m 08s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/2/gate/gate-tempest-dsvm-neutron-src-oslo.utils/752b8bc : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"21f5d3638b9228100c7ccfe3046cc95899abe571","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2014-10-03 23:53:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"08d192fd9724c16a3f4c2047b559682f72707a94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-04 00:55:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-pep8/c2c41b1 : SUCCESS in 43s\n- gate-oslo.utils-docs http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-docs/e2c4832 : FAILURE in 46s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python26/a622413 : SUCCESS in 59s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python27/77c03a9 : SUCCESS in 1m 12s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/2/check/gate-oslo.utils-python33/81eab29 : SUCCESS in 2m 38s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/2/check/gate-tempest-dsvm-src-oslo.utils/94504de : SUCCESS in 57m 10s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/2/check/gate-tempest-dsvm-neutron-src-oslo.utils/af8723f : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7e94d552692a717b1853dd8cce7e8883df914b88","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-10-04 12:56:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da93591158ac6b91fcabbc69e829eb8b59b8d3d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-04 13:48:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-pep8/cf6bcb8 : SUCCESS in 29s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/3/check/gate-oslo.utils-docs/1f16780/doc/build/html/ : SUCCESS in 31s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python26/1e5b3b6 : SUCCESS in 57s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python27/75728dc : SUCCESS in 42s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python33/b861b2d : SUCCESS in 2m 18s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-src-oslo.utils/38b44fd : SUCCESS in 41m 40s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-neutron-src-oslo.utils/b332bbc : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":3},{"id":"c9c60a9a5f8373f8af98f7cdeb2a318ee44307a9","author":{"_account_id":7491,"name":"Viktor Serhieiev","email":"viktor.sergeyev@gmail.com","username":"vsergeyev"},"date":"2014-10-06 09:25:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f6e87d335c40fae1278ee9398925849948748552","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-10-07 12:52:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1f856c569e05f4496b9cc99a0a32e2d328d1d7ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-07 14:13:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-pep8/b9b1f13 : SUCCESS in 42s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/3/check/gate-oslo.utils-docs/551292c/doc/build/html/ : SUCCESS in 54s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python26/804c552 : FAILURE in 3m 28s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python27/cb13cc4 : SUCCESS in 1m 06s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python33/b18629f : SUCCESS in 2m 35s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-src-oslo.utils/33522df : SUCCESS in 57m 23s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-neutron-src-oslo.utils/cdf6afa : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":3},{"id":"0c7af3c585bf4e7fa0c354e0787917abafde1355","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-10-07 17:49:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b766f10056cc65c19eb6d0146e779e9aaf711910","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-07 19:29:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-pep8/17a56ba : SUCCESS in 49s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/3/check/gate-oslo.utils-docs/e18c314/doc/build/html/ : SUCCESS in 1m 20s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python26/75edda8 : SUCCESS in 1m 46s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python27/d73c1c5 : SUCCESS in 1m 12s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/3/check/gate-oslo.utils-python33/25a4ac9 : SUCCESS in 3m 22s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-src-oslo.utils/ce9f6dd : SUCCESS in 45m 28s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/3/check/gate-tempest-dsvm-neutron-src-oslo.utils/469b534 : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":3},{"id":"6b91f8e3baf7dc6801cdb010b17d9b4a4879a0f9","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-10-27 18:13:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"80271548a861c3ec1d10d78427c8e22364ece627","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-27 21:05:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.utils-pep8 http://logs.openstack.org/97/119097/4/check/gate-oslo.utils-pep8/6fc2739 : SUCCESS in 24s\n- gate-oslo.utils-docs http://docs-draft.openstack.org/97/119097/4/check/gate-oslo.utils-docs/9f92edc/doc/build/html/ : SUCCESS in 1m 19s\n- gate-oslo.utils-python27 http://logs.openstack.org/97/119097/4/check/gate-oslo.utils-python27/fa089f3 : SUCCESS in 41s\n- gate-oslo.utils-python33 http://logs.openstack.org/97/119097/4/check/gate-oslo.utils-python33/e5cf2aa : SUCCESS in 2m 39s\n- gate-tempest-dsvm-src-oslo.utils http://logs.openstack.org/97/119097/4/check/gate-tempest-dsvm-src-oslo.utils/d6b17b5 : SUCCESS in 55m 18s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/97/119097/4/check/gate-tempest-dsvm-neutron-src-oslo.utils/c133e4a : SUCCESS in 1h 05m 18s\n- gate-oslo.utils-python26 http://logs.openstack.org/97/119097/4/check/gate-oslo.utils-python26/e5bd334 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4},{"id":"814f89824e37ec449962d2ebe87103d94cb02e87","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-11-07 10:23:09.000000000","message":"Abandoned\n\nConsensus is against auto-generation","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ad3fd54d14ef139dabac02d1e706d4d0791bcc98","revisions":{"f6811ddf2c72b5f61726784dffc7fc5298f21a40":{"kind":"REWORK","_number":1,"created":"2014-09-04 15:53:04.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/97/119097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/97/119097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/1"}}},"commit":{"parents":[{"commit":"281df4b601a5a532353655a7e960dacb5cab0e25","subject":"Merge \"Just use int(BOOL) to convert to 1 or 0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/281df4b601a5a532353655a7e960dacb5cab0e25"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-04 15:52:57.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-04 15:52:57.000000000","tz":-240},"subject":"Auto generate the API documentation","message":"Auto generate the API documentation\n\nChange-Id: I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f6811ddf2c72b5f61726784dffc7fc5298f21a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f6811ddf2c72b5f61726784dffc7fc5298f21a40"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"281df4b601a5a532353655a7e960dacb5cab0e25","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5f2962600d5149ca2487eeec8202511aa039e743":{"kind":"REWORK","_number":2,"created":"2014-09-05 01:34:00.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/97/119097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/97/119097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/2"}}},"commit":{"parents":[{"commit":"281df4b601a5a532353655a7e960dacb5cab0e25","subject":"Merge \"Just use int(BOOL) to convert to 1 or 0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/281df4b601a5a532353655a7e960dacb5cab0e25"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-04 15:52:57.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-05 01:33:30.000000000","tz":-240},"subject":"Auto generate the API documentation","message":"Auto generate the API documentation\n\nOnce we use the pbr based build to generate files, we prune\nthe indexes to remove references to code we would like to\nhide like the common modules and private modules in conf.py.\nWe reuse the same excluded_patterns variable that is already\nused in Sphinx and filter these modules from autoindex.rst\n\nChange-Id: I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5f2962600d5149ca2487eeec8202511aa039e743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5f2962600d5149ca2487eeec8202511aa039e743"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"281df4b601a5a532353655a7e960dacb5cab0e25","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a1eb7236779161c4e4a0312327b1048882f6f256":{"kind":"REWORK","_number":3,"created":"2014-10-04 12:56:41.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/97/119097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/97/119097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/3"}}},"commit":{"parents":[{"commit":"5ef07d9b4d35c5123a4c259668b08e55079f72ad","subject":"Merge \"Imported Translations from Transifex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5ef07d9b4d35c5123a4c259668b08e55079f72ad"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-04 15:52:57.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-10-04 12:56:34.000000000","tz":-240},"subject":"Auto generate the API documentation","message":"Auto generate the API documentation\n\nOnce we use the pbr based build to generate files, we prune\nthe indexes to remove references to code we would like to\nhide like the common modules and private modules in conf.py.\nWe reuse the same excluded_patterns variable that is already\nused in Sphinx and filter these modules from autoindex.rst\n\nChange-Id: I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a1eb7236779161c4e4a0312327b1048882f6f256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a1eb7236779161c4e4a0312327b1048882f6f256"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ef07d9b4d35c5123a4c259668b08e55079f72ad","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ad3fd54d14ef139dabac02d1e706d4d0791bcc98":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2014-10-27 18:13:48.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/97/119097/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/97/119097/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/97/119097/4"}}},"commit":{"parents":[{"commit":"7b32a91acc86cf26aa64344767f553dff27f34bf","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/7b32a91acc86cf26aa64344767f553dff27f34bf"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2014-09-04 15:52:57.000000000","tz":-240},"committer":{"name":"Davanum Srinivas (dims)","email":"davanum@gmail.com","date":"2014-10-27 18:13:48.000000000","tz":0},"subject":"Auto generate the API documentation","message":"Auto generate the API documentation\n\nOnce we use the pbr based build to generate files, we prune\nthe indexes to remove references to code we would like to\nhide like the common modules and private modules in conf.py.\nWe reuse the same excluded_patterns variable that is already\nused in Sphinx and filter these modules from autoindex.rst\n\nChange-Id: I21e516a80f20eb2d5965a5a63c7dfd2d231c67e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/ad3fd54d14ef139dabac02d1e706d4d0791bcc98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/ad3fd54d14ef139dabac02d1e706d4d0791bcc98"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7b32a91acc86cf26aa64344767f553dff27f34bf","is_merged_in_target_branch":true,"change_id":"Ibc523c0034585230f072919fbab576efc4d32f35","change_number":127783,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
