)]}'
{"id":"openstack%2Foslo.utils~328149","triplet_id":"openstack%2Foslo.utils~master~Ia3a81f3767428c153e1daf25e26da3a39031d14b","project":"openstack/oslo.utils","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ia3a81f3767428c153e1daf25e26da3a39031d14b","subject":"Imported Translations from Zanata","status":"ABANDONED","created":"2016-06-10 08:06:36.000000000","updated":"2016-06-10 18:45:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8e49a4747349bd5d6a7ccbf77f5e485a1e9a306","_number":328149,"virtual_id_number":328149,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-06-10 09:29:04.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2016-06-10 08:07:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-10 08:07:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2016-06-10 09:29:04.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"34196bb9e1cabad057bc478ac284e49b612f69e1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-06-10 08:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7efc3a6d2a6df3ae3caa2f5d7e4239855a4ae65","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2016-06-10 08:07:42.000000000","message":"Patch Set 1: Code-Review-1\n\nBroken UTF-8 strings, let\u0027s fix infra setup first.","accounts_in_message":[],"_revision_number":1},{"id":"b08994779f113606d66d6666442251e136bcf539","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-10 09:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.utils-docs http://docs-draft.openstack.org/49/328149/1/check/gate-oslo.utils-docs/dea0a14//doc/build/html/ : SUCCESS in 5m 21s\n- gate-oslo.utils-pep8 http://logs.openstack.org/49/328149/1/check/gate-oslo.utils-pep8/5845fc4/ : SUCCESS in 3m 40s\n- gate-oslo.utils-python27 http://logs.openstack.org/49/328149/1/check/gate-oslo.utils-python27/51d95de/ : SUCCESS in 5m 23s\n- gate-oslo.utils-python34 http://logs.openstack.org/49/328149/1/check/gate-oslo.utils-python34/7dd1c97/ : SUCCESS in 3m 53s\n- gate-tempest-dsvm-neutron-src-oslo.utils http://logs.openstack.org/49/328149/1/check/gate-tempest-dsvm-neutron-src-oslo.utils/a58f595/ : SUCCESS in 1h 20m 08s\n- gate-tempest-dsvm-neutron-src-oslo.utils-mitaka http://logs.openstack.org/49/328149/1/check/gate-tempest-dsvm-neutron-src-oslo.utils-mitaka/0fc676c/ : SUCCESS in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaa146b13d7c752babeec29b64630257ae963244","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2016-06-10 18:45:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e572758c658bdb0ef620743b66e65b251c6b7e4d","revisions":{"e572758c658bdb0ef620743b66e65b251c6b7e4d":{"kind":"REWORK","_number":1,"created":"2016-06-10 08:06:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/328149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/49/328149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/49/328149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/49/328149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/49/328149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/49/328149/1"}}},"commit":{"parents":[{"commit":"6439ee0ab5ad876e664d77286d64b1da82fc0b00","subject":"Merge \"Avoid catching generic exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/6439ee0ab5ad876e664d77286d64b1da82fc0b00"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-10 08:06:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-10 08:06:35.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://wiki.openstack.org/wiki/Translations/Infrastructure\n\nChange-Id: Ia3a81f3767428c153e1daf25e26da3a39031d14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/e572758c658bdb0ef620743b66e65b251c6b7e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/e572758c658bdb0ef620743b66e65b251c6b7e4d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6439ee0ab5ad876e664d77286d64b1da82fc0b00","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
