)]}'
{"id":"openstack%2Foslo.utils~924766","triplet_id":"openstack%2Foslo.utils~master~I999f34acc3b7a67db5e2c7ee241e661ce688e5fe","project":"openstack/oslo.utils","branch":"master","topic":"add-format-inspector","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-08-07 10:42:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I999f34acc3b7a67db5e2c7ee241e661ce688e5fe","subject":"Refactor some things for oslo","status":"MERGED","created":"2024-07-23 17:16:00.000000000","updated":"2024-08-07 10:44:17.000000000","submitted":"2024-08-07 10:42:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924766-add-format-inspector","meta_rev_id":"f979d47451b364db24febd77ede41311fb2e4a0f","_number":924766,"virtual_id_number":924766,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 10:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-02 14:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-31 17:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-06 14:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-06 14:05:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 18:58:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 14:12:14.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-02 14:26:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-06 14:05:51.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"4cb567129b549fe63fd91e8f7050bb2200fcb29d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-23 17:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57d77ad694afea8cb9dd33c3ef194f78a1611ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 18:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe790851dd5495997f9a077b0f90581\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ae39ae95a9b4baabc2cc84469e9aba1 : SUCCESS in 1h 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea09067d14674e0d9a246981e17c6344 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/081d264320fc477595882ec0f0d871bc : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a70c771478745899b8c769053e1690a : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3571048d58540e88678f64d5c732984 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4502040753f491ebfab53a81f68d755 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c6423c0236d44e6d7cbfb3aa899700120aa713b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-29 13:53:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"59a774495d92d048cf5b0f447da7d74df02dacf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:41:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7216e23b638c4db89f84a82cab46c6a8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5942417b9c942d19ba550e9df853d84 : SUCCESS in 1h 46m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd74b3643ec14aec8af43e121b3ba592 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c0649763fd40779b34dcfe69aeb5b1 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fbae09809b247f9be947e5482752fc2 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b1111b4e0234f6ca69ea28ead5019e3 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc4c3a04172545aaa24c78eec4fcacd7 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90c389ce02b4c12b239aa22b7ec7e18 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"db4b70f9d9c3ee8bae58fc605dd2e602c78c5bf6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-31 14:12:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dce3b7f64fd9832f4e5e8a9689928a86881e70b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-31 16:00:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"42c5a5e56cd1f0cb1907360e20cefd57fb481393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:27:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e835a73ce24c998eaf2f8cfdf10c61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b8f699ec5e5477ebae5d442d2696800 : SUCCESS in 1h 25m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5dbdf5e2bcd46e394bbe71388243d7a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc375f226d7645d488168b985a973bb9 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/220eb4cef52c41e390010b475fe117a9 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58ec27c27d6f4495b085d24ce1f92727 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac4ee7e9177d4579b7555dad7ea90594 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570508eeb52f48c590ecb7449ab25bb6 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5c7b0994ae062b39f0865e713c270099b31a927c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-31 17:40:45.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e69a40eff7cb1da844338306c6c53db232048529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 19:24:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90352d2e8e7d4e1badcccd9e1812b7e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5089eaeb357b45089729afe897824cbc : SUCCESS in 1h 37m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5c65e63e03f47f880a4e34fe79ad8f5 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5c50b9068845f5a2ecf11715a9bec1 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a3d9006eaae472b8f6adbe8464227a1 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a85908719763472588bb448722bafbf1 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09f601cbf4b347f6acee776ed3051a78 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5df58b4e9b6485e98653f276ae35619 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"0c924f4a04a1fdda58d3e0d0223b797560a806ea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-02 14:26:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5ce5d2638797882242c97fa178fb3123fdc132cc","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-06 14:05:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c22084c0231309b5c04a95f84ebec2374fc7aac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 08:52:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3da28f39f50993d6dec9762530270c47f2344dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:42:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71714c0be1dd4a35955d98136456bf35\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed938d94931b4f7b841ec4bb772589c5 : SUCCESS in 1h 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/401f54b2f86a4ea9b6379c2640646dfd : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac147272d59f460f9aca86fc82deaecf : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d66728c52754b93a1aaba81082bcf2e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7f539a26c54bb5aa4522331e4070a4 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":4},{"id":"98d4ad6c81e3346d936ca577fe748952a948e20f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:42:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f979d47451b364db24febd77ede41311fb2e4a0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:44:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49203394101d45bfb15af488099f8e60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99307194251144b5b5466a8944bde26f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f756b8645ec628202bbcf47c8c1064e247ecb07","revisions":{"f1f0e2f706961450fc7202c4b72dc2dff3321467":{"kind":"REWORK","_number":1,"created":"2024-07-23 17:16:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/924766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/66/924766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/1"}}},"commit":{"parents":[{"commit":"d6d1915b41e9d7a5b8c4740b95a5b48d7c845a79","subject":"Support VMDK sparse with footer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/d6d1915b41e9d7a5b8c4740b95a5b48d7c845a79"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-23 17:12:18.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-23 17:13:48.000000000","tz":-420},"subject":"Refactor some things for oslo","message":"Refactor some things for oslo\n\nThis reflects some miscellaneous feedback around suitability for oslo:\n\n1. Removes the TraceDisabled helper in favor of a simple boolean\n2. Renames chunked_reader() to make it \"private\" since it is only\n   intended to be used internally.\n\nChange-Id: I999f34acc3b7a67db5e2c7ee241e661ce688e5fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f1f0e2f706961450fc7202c4b72dc2dff3321467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f1f0e2f706961450fc7202c4b72dc2dff3321467"}]},"branch":"refs/heads/master"},"2def8022b5b0862ec4bd2861d9c8cda5b7cc9673":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-29 13:53:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/924766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/66/924766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/2"}}},"commit":{"parents":[{"commit":"1b75c8c60b53580f40198b1c093be3ddaac3bdb4","subject":"Support VMDK sparse with footer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1b75c8c60b53580f40198b1c093be3ddaac3bdb4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-23 17:12:18.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-29 13:53:16.000000000","tz":-420},"subject":"Refactor some things for oslo","message":"Refactor some things for oslo\n\nThis reflects some miscellaneous feedback around suitability for oslo:\n\n1. Removes the TraceDisabled helper in favor of a simple boolean\n2. Renames chunked_reader() to make it \"private\" since it is only\n   intended to be used internally.\n\nChange-Id: I999f34acc3b7a67db5e2c7ee241e661ce688e5fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/2def8022b5b0862ec4bd2861d9c8cda5b7cc9673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/2def8022b5b0862ec4bd2861d9c8cda5b7cc9673"}]},"branch":"refs/heads/master"},"a02ee5d2266443cdb381175f90726c4064f5f1ba":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-31 16:00:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/924766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/66/924766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/3"}}},"commit":{"parents":[{"commit":"a5b33a4f40127176c23324d1e0bd7a3ed99f7765","subject":"Support VMDK sparse with footer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a5b33a4f40127176c23324d1e0bd7a3ed99f7765"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-23 17:12:18.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-31 15:59:37.000000000","tz":-420},"subject":"Refactor some things for oslo","message":"Refactor some things for oslo\n\nThis reflects some miscellaneous feedback around suitability for oslo:\n\n1. Removes the TraceDisabled helper in favor of a simple boolean\n2. Renames chunked_reader() to make it \"private\" since it is only\n   intended to be used internally.\n\nChange-Id: I999f34acc3b7a67db5e2c7ee241e661ce688e5fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a02ee5d2266443cdb381175f90726c4064f5f1ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a02ee5d2266443cdb381175f90726c4064f5f1ba"}]},"branch":"refs/heads/master"},"1f756b8645ec628202bbcf47c8c1064e247ecb07":{"kind":"NO_CHANGE","_number":4,"created":"2024-07-31 17:40:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/66/924766/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/66/924766/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/66/924766/4"}}},"commit":{"parents":[{"commit":"cfc0e1039a9862100165af2dfdceb002a7a5c0e7","subject":"Support VMDK sparse with footer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/cfc0e1039a9862100165af2dfdceb002a7a5c0e7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-23 17:12:18.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-31 17:40:36.000000000","tz":-420},"subject":"Refactor some things for oslo","message":"Refactor some things for oslo\n\nThis reflects some miscellaneous feedback around suitability for oslo:\n\n1. Removes the TraceDisabled helper in favor of a simple boolean\n2. Renames chunked_reader() to make it \"private\" since it is only\n   intended to be used internally.\n\nChange-Id: I999f34acc3b7a67db5e2c7ee241e661ce688e5fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1f756b8645ec628202bbcf47c8c1064e247ecb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1f756b8645ec628202bbcf47c8c1064e247ecb07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
