)]}'
{"id":"openstack%2Foslo.utils~924859","triplet_id":"openstack%2Foslo.utils~master~I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa","project":"openstack/oslo.utils","branch":"master","topic":"add-format-inspector","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-08-07 15:40:05.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-08-07 15:40:05.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-08-07 15:40:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa","subject":"Fix qcow2 feature flag checks (for the future)","status":"MERGED","created":"2024-07-24 14:29:23.000000000","updated":"2024-08-07 15:43:07.000000000","submitted":"2024-08-07 15:40:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"924859-add-format-inspector","meta_rev_id":"47781fbbe9b338efd983d17a19c137fa48df82ee","_number":924859,"virtual_id_number":924859,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 15:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-31 17:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-08-06 16:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-08-07 12:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-06 14:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-06 14:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 14:51:28.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-07-24 16:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-25 09:49:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-07-31 14:14:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-06 14:59:28.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-08-06 16:50:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e2fa7f1036d62cedb4fe69317e6fd9b167cd8490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-24 14:29:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9ac82db67b0dbeff87dd051d893d5bd986fce14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-24 14:47:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fc823558f0cb1f0674d1d6a389d7511aa8b6514","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-24 14:51:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ce6ae3663d36eb18ef32a3136226cad083d2af94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 16:06:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d075e80dd374a6fa9bf3dd47120c949\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd07c1a02b264208827460d8bb697724 : SUCCESS in 1h 18m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6647abe7d3b4e65b1f3975fe2ee8375 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b7bb3abb0f433baddca7097b8db65b : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1491663e9c9497cb14d5134ca3fbeac : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48b56568b77445838d0a1cce31a016a8 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f20c17e645cf4a258dc95e1efbaff0f5 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090f38bebd404c409b49ba1023b98877 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f6e5209f37cdce7f51b34f5bf1babf32d045696f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-25 09:49:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"da9348b50e2fcecb3b5632c64a24afd8f68c6344","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-25 13:39:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"486cf0ee7163fd90abb5daa53f36a80cfd9377d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-29 13:53:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bc6a7fab57368e4d38461b901d404920dd5c6d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:29:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba6ff544275467194392c95a72285cf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97685a289285417dbd05a665d8a112e6 : SUCCESS in 1h 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045edd11458c4cb89775eac8ab48fe0d : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8a9db17a8774989815b4048f2047033 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8b071ed8d064f7a96a8a3d665b477ae : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fa47c90eda143f2830e23fd8af0ab0b : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e337c7b17c9149fba72e9fbb11b83e0f : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3c67a961b44a9b9b382833f50c4807 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f52ddbd2bd20e616bf0c612b553d91400c25ef1d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-31 14:14:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"efc13af70eec5e6a58fdf859b9775d7595759fc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-31 16:00:33.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c45cbc9b898c2f09a6384c722bf4cabaccc5c041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:05:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10355ce80e4342f0a8260c497527f394\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2ebf84732574cc5b1d4b1732089ad67 : FAILURE in 1h 03m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5867e5c70e48eb8aa41d09e6cc18a3 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb69edae78d459c8cf93fb1ceb3ff88 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c0a9074bb4045efbd580a72af9239e1 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9af96889ec9a48eeb03bfbc59140d758 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/340e8381c5ec4af18dcb49e39057641d : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c406248c0f35441f8e4828d5078b350a : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":4},{"id":"cf05b8472b11b76862105b0239235855f64d048c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-31 17:40:45.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0d816bb1b77e73a474f76ad096becb6f293bda75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 19:25:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ed5809da1342bda0884d3f6e01c180\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a00c50da6c3741e19746110bba0af3ed : SUCCESS in 1h 39m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd8d9ca8fca74715aa1fed7d556f7dc3 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0fb065e1ce4326b2b13dd5ab18d0ae : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aad5e878bfc64508ade6019da5afffaa : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/151bc7ba55074652babecbcdec0a212b : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0babb020f8234b7eb2f0f67735bb36f4 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58e65f2565740618f59fdc2b967a259 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":5},{"id":"33666411cf6977468c4d1bbc919e93d2b5be59b1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-06 14:59:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"75cdc20a8e8e3d22eed95683e2df94beea1e8361","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-06 16:50:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"04b1ade71f9844637175f044f067adcdea32d33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 08:52:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2a77bb379036b4c30612a3522c5f6fb1a4d6c36b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:42:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aeaf10d965d49f7b83092874294b074\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/335df69e4f7e4818bd2f39986a0a3744 : SUCCESS in 1h 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cd27531d1f4a9889cd6d6a07ed0e52 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00aa9ded13454fca92cfbf343ca61010 : FAILURE in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a69adec71b74d46867bb8f624d1cfa4 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d027366f4d164d01b39c24b14c5dd920 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":5},{"id":"691832f30cb859dbad1c316542bba82778758196","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-07 12:11:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71150adbe16ac2c8932c7a5062b396290f458657","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 13:57:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c574a8e0694694806cd3a82594b352\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f129f3df86a141bdadc4b36543627874 : SUCCESS in 1h 45m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c25d6fdae14ce2b69adef5e434b7c6 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3c730b0df54202bd30158ffc8ba6d7 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10743185935641e48438ba504ffc4c37 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c7ad4c14ab2457aab05afb55e316eed : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aee2ec98d9494b03afceb997fc2e1694 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8923824ce5a4c2a81f3ad2ade286b8f : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":5},{"id":"b0799a51a3d97ef5f21da4956989ed05cf53f490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 13:57:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"52aea3914c83ea5141450ffa381a12506b4039a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 15:40:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b2b2054e7d4b9c91f70d3c2e52d238\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2191b9ca0d2f46d7884ddcc9900285bf : SUCCESS in 1h 37m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c492ce8b4f644889c0da24bcad58fc3 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe4fcaae4914cacaec2cd9a06429271 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bbdd35596604e009bc18416d0942449 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f09e259ebf458dbe4c84bc2ac41769 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"96fbf3b169709bedb4305ba94bd68a918d8423a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 15:40:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"47781fbbe9b338efd983d17a19c137fa48df82ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 15:43:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cffdd5ac065145de8e27226e0d4c8ba6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b5600ca3834b70a178945cc5710656 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1290708dbb4f56375c7563db9ffc7ff58f81d6ba","revisions":{"ada637481ff56e922d202cadfcd77580a1ca80c4":{"kind":"REWORK","_number":1,"created":"2024-07-24 14:29:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/59/924859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/59/924859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/1"}}},"commit":{"parents":[{"commit":"f1f0e2f706961450fc7202c4b72dc2dff3321467","subject":"Refactor some things for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f1f0e2f706961450fc7202c4b72dc2dff3321467"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"subject":"Fix qcow2 feature flag checks (for the future)","message":"Fix qcow2 feature flag checks (for the future)\n\nThis is not a problem now, but in a hypothetical future where qcow2\nhas grown more than double the features it has added since inception,\nwe need to be able to check feature bits \u003e8. Due to an oversight in\nthe feature flag mask calculation, we may do that incorrectly today.\n\nSo close that up now to avoid the issue later. Also, avoid feature\nflag checking at all if the internal format version is \"2\", which\ndid not have the flags, and also fail the check if the version is not\none of the two we know of (i.e. 2 or 3).\n\nChange-Id: I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa\nRelated-Bug: #2073413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/ada637481ff56e922d202cadfcd77580a1ca80c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/ada637481ff56e922d202cadfcd77580a1ca80c4"}]},"branch":"refs/heads/master"},"b4d03c2bc09c794e5ea07bcfed25e6df72dadf34":{"kind":"REWORK","_number":2,"created":"2024-07-24 14:47:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/59/924859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/59/924859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/2"}}},"commit":{"parents":[{"commit":"f1f0e2f706961450fc7202c4b72dc2dff3321467","subject":"Refactor some things for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f1f0e2f706961450fc7202c4b72dc2dff3321467"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:47:06.000000000","tz":-420},"subject":"Fix qcow2 feature flag checks (for the future)","message":"Fix qcow2 feature flag checks (for the future)\n\nThis is not a problem now, but in a hypothetical future where qcow2\nhas grown more than double the features it has added since inception,\nwe need to be able to check feature bits \u003e8. Due to an oversight in\nthe feature flag mask calculation, we may do that incorrectly today.\n\nSo close that up now to avoid the issue later. Also, avoid feature\nflag checking at all if the internal format version is \"2\", which\ndid not have the flags, and also fail the check if the version is not\none of the two we know of (i.e. 2 or 3).\n\nChange-Id: I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa\nRelated-Bug: #2073413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/b4d03c2bc09c794e5ea07bcfed25e6df72dadf34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/b4d03c2bc09c794e5ea07bcfed25e6df72dadf34"}]},"branch":"refs/heads/master"},"5dad87d2fc1af0c6d1d5a4d219cc20107a6e221d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-29 13:53:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/59/924859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/59/924859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/3"}}},"commit":{"parents":[{"commit":"2def8022b5b0862ec4bd2861d9c8cda5b7cc9673","subject":"Refactor some things for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/2def8022b5b0862ec4bd2861d9c8cda5b7cc9673"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-29 13:53:16.000000000","tz":-420},"subject":"Fix qcow2 feature flag checks (for the future)","message":"Fix qcow2 feature flag checks (for the future)\n\nThis is not a problem now, but in a hypothetical future where qcow2\nhas grown more than double the features it has added since inception,\nwe need to be able to check feature bits \u003e8. Due to an oversight in\nthe feature flag mask calculation, we may do that incorrectly today.\n\nSo close that up now to avoid the issue later. Also, avoid feature\nflag checking at all if the internal format version is \"2\", which\ndid not have the flags, and also fail the check if the version is not\none of the two we know of (i.e. 2 or 3).\n\nChange-Id: I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa\nRelated-Bug: #2073413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5dad87d2fc1af0c6d1d5a4d219cc20107a6e221d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/5dad87d2fc1af0c6d1d5a4d219cc20107a6e221d"}]},"branch":"refs/heads/master"},"72ac1c78c16d942cbea7991b640c3a71325499cd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-31 16:00:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/59/924859/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/59/924859/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/4"}}},"commit":{"parents":[{"commit":"a02ee5d2266443cdb381175f90726c4064f5f1ba","subject":"Refactor some things for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a02ee5d2266443cdb381175f90726c4064f5f1ba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-31 15:59:37.000000000","tz":-420},"subject":"Fix qcow2 feature flag checks (for the future)","message":"Fix qcow2 feature flag checks (for the future)\n\nThis is not a problem now, but in a hypothetical future where qcow2\nhas grown more than double the features it has added since inception,\nwe need to be able to check feature bits \u003e8. Due to an oversight in\nthe feature flag mask calculation, we may do that incorrectly today.\n\nSo close that up now to avoid the issue later. Also, avoid feature\nflag checking at all if the internal format version is \"2\", which\ndid not have the flags, and also fail the check if the version is not\none of the two we know of (i.e. 2 or 3).\n\nChange-Id: I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa\nRelated-Bug: #2073413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/72ac1c78c16d942cbea7991b640c3a71325499cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/72ac1c78c16d942cbea7991b640c3a71325499cd"}]},"branch":"refs/heads/master"},"1290708dbb4f56375c7563db9ffc7ff58f81d6ba":{"kind":"NO_CHANGE","_number":5,"created":"2024-07-31 17:40:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/59/924859/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/59/924859/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/59/924859/5"}}},"commit":{"parents":[{"commit":"1f756b8645ec628202bbcf47c8c1064e247ecb07","subject":"Refactor some things for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1f756b8645ec628202bbcf47c8c1064e247ecb07"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-24 14:26:34.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-31 17:40:36.000000000","tz":-420},"subject":"Fix qcow2 feature flag checks (for the future)","message":"Fix qcow2 feature flag checks (for the future)\n\nThis is not a problem now, but in a hypothetical future where qcow2\nhas grown more than double the features it has added since inception,\nwe need to be able to check feature bits \u003e8. Due to an oversight in\nthe feature flag mask calculation, we may do that incorrectly today.\n\nSo close that up now to avoid the issue later. Also, avoid feature\nflag checking at all if the internal format version is \"2\", which\ndid not have the flags, and also fail the check if the version is not\none of the two we know of (i.e. 2 or 3).\n\nChange-Id: I352d6eceff2a90dd4a29d9eeeed1360f7b02dafa\nRelated-Bug: #2073413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1290708dbb4f56375c7563db9ffc7ff58f81d6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/1290708dbb4f56375c7563db9ffc7ff58f81d6ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
