)]}'
{"id":"openstack%2Foslo.utils~926651","triplet_id":"openstack%2Foslo.utils~master~I96fd8e6bec846065c9dc73d2f3af5899099461a4","project":"openstack/oslo.utils","branch":"master","topic":"remove-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-01 12:23:30.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-10-01 12:23:30.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-10-01 12:23:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96fd8e6bec846065c9dc73d2f3af5899099461a4","subject":"Remove Python 3.8 support","status":"MERGED","created":"2024-08-20 14:24:00.000000000","updated":"2024-10-01 12:24:31.000000000","submitted":"2024-10-01 12:23:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926651-remove-py38","meta_rev_id":"2352fe0c334870f495f8d82a4116ffacd15cfb72","_number":926651,"virtual_id_number":926651,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-01 12:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-10-01 10:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-09-30 08:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-10-01 10:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 14:37:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-20 16:03:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 13:34:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-08-27 14:49:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-10-01 10:58:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"fa2dd2da483f87b36318357ae755bd4d8a790933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-20 14:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bcb505620c87006a20f37fe2f22c1f3407b857e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 14:37:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43cc9d2bc25556de37d095bbd303dfee1c66cb88","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-20 14:49:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e7c25f30583e85d21880b4a24f82e64f0710a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:03:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff87e5ee17f041a3bd6dab2c9c378a00\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3eb2fb0840a3455ba0bed143dcd7961b : SUCCESS in 2m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/df8308be452245e4ae7a1e30497566bf : SUCCESS in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db788c8a908d4bcda04b1c05157400a3 : SUCCESS in 1h 34m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b159604b74c4ee2b60e9b660305d999 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87942eedccea4baea1d79bd2432a70fd : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/053652de80e24b65a9315f96377e0bbe : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3887e740f3af4613847dde75824e7724 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62a6bf7c41744f11b9bb1f89d268424e : FAILURE in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d204a656b854825a3699e3dc647ae24 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2f896fab096d914289e79639a8808937983dbb1f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-26 13:34:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42e3c9cad02676a2bdb00f26deffc41f655b2964","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-26 13:47:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c96789e0956c837053b0d5d53b88a8a18f72a8a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 11:09:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe8a38a2f98a2acd1aa2534f92e899b68b1a9923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 11:11:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67003bcb2a6c7407bc641588caeb192dbe73a4e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 12:47:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c246b14524b445f58a8e68d8f48f9d4b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4e45c6a8da9d4335b41db890dcd4a7aa : SUCCESS in 1m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d1110316a069419ca5d91d349edb8b45 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64493329fb45408589a6933152328b75 : SUCCESS in 1h 34m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d569d017965246fab82799c8dad854d7 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ecd172f5224557a72ce5eafe6bb954 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41713caba24d4546a0abff76d4085b80 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/579bed8c50ac4ed8a920cedaa828cb4a : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d99cac1569643358de2184d5a880032 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc212c12f84b4e46b1b49d06a9e567ec : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"78013f85f4916333959e6e7988290ff240f63a52","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 12:58:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e34c69ecd7946b5d41fe5da946e6cb251345efb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-27 14:49:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b5afa1529a9cb40e2194591df7521bdbf009c256","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-09-30 07:51:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"274310d592da41f24798910db4864cf668d4b983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-09-30 07:54:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b8f252d9ed13297a9d4957452654cc7e268056d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 08:25:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4ba3d725a626e86e59837c8f8d4c202b0d2565a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 08:29:02.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b762cb86ba40574eadf66c5700a7bfc74e8ccbb0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 08:38:14.000000000","message":"Topic remove_py38 removed","accounts_in_message":[],"_revision_number":5},{"id":"454c788efa06784c321b9aa588e508c37f95abd1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 08:38:20.000000000","message":"Topic set to remove-py38","accounts_in_message":[],"_revision_number":5},{"id":"f9c3f1d01e02f934ada5e885192c237343b4c1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 09:30:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5124ac94345452da11dd9e7b05d948c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef1004853e704d09b657e818fc4d94e5 : SUCCESS in 1m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/89038a489ece4c1eb7c7fc248afda9e7 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b08ad43b2ec45e99247f9e32a93e446 : SUCCESS in 1h 03m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51931e30d7854f1e9111eaf321996fda : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d19036ca2a54980947c34bab3d16fa1 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a26b8206f274cdf908fba7fb1c244e0 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e36ed05a54b41f691561589b77c6112 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148dcd39cad543319d70cccdfa338aad : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbfc86ab1e6440fd91758013d450afea : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":5},{"id":"a55c583b272684bcbb055e8e9f93886e725a3697","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-01 10:58:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b5a5c9c40aab4659a1c1b1c592f34eae10993b44","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-01 10:58:26.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"984f31e7e1e9bfae01016b05d7a9c0beffb65480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 10:58:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f66a584a8b7474415f3cbef733f893a8c93e43af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:23:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429ebd1ef65246c8b6380b3f3c36020e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0412999bdb274eec83a1d1ed7342430c : SUCCESS in 1m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f318dcd1de9403a99dc93803577b844 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/251e812ef49843b0b4d0c52d0c9f666b : SUCCESS in 1h 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9b302f7ead43a4a96501953eb0ee29 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8266ccfcb974492083fe1666ab2381de : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc40d7d283934c72b08171c2bb427815 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd364de78764555b69b8b25113efc12 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f01c981d3b4421f95ecef3eb900dcb3 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":5},{"id":"13b5ba1887f18be4e06ff55e1916692af6f81fd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:23:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2352fe0c334870f495f8d82a4116ffacd15cfb72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 12:24:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3349d5ab837e4632bb36b8f7857b9284\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616936bde4864987ad4b91724ab3b7b1 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e774757719a46929eca4e04318c64a3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fa2ecb269a4aaccfbb4903d95e8f13042d83aece","revisions":{"fb99b6fdeda7165d22fb2df8d212cddba6869019":{"kind":"REWORK","_number":1,"created":"2024-08-20 14:24:00.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/51/926651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/51/926651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/1"}}},"commit":{"parents":[{"commit":"4e596cd865c0e23f98fab47d09e1cb31ed51b14e","subject":"Add release note for format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/4e596cd865c0e23f98fab47d09e1cb31ed51b14e"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:58.000000000","tz":120},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I96fd8e6bec846065c9dc73d2f3af5899099461a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/fb99b6fdeda7165d22fb2df8d212cddba6869019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/fb99b6fdeda7165d22fb2df8d212cddba6869019"}]},"branch":"refs/heads/master"},"78a76099ab03d7e108f6637e7e694e19b20a1ce0":{"kind":"REWORK","_number":2,"created":"2024-08-27 11:09:55.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/51/926651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/51/926651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/2"}}},"commit":{"parents":[{"commit":"4e596cd865c0e23f98fab47d09e1cb31ed51b14e","subject":"Add release note for format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/4e596cd865c0e23f98fab47d09e1cb31ed51b14e"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-27 11:09:45.000000000","tz":120},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I96fd8e6bec846065c9dc73d2f3af5899099461a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/78a76099ab03d7e108f6637e7e694e19b20a1ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/78a76099ab03d7e108f6637e7e694e19b20a1ce0"}]},"branch":"refs/heads/master"},"f6dbd116fd0891821cf37647eb7e10b490164328":{"kind":"REWORK","_number":3,"created":"2024-08-27 11:11:43.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/51/926651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/51/926651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/3"}}},"commit":{"parents":[{"commit":"4e596cd865c0e23f98fab47d09e1cb31ed51b14e","subject":"Add release note for format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/4e596cd865c0e23f98fab47d09e1cb31ed51b14e"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-27 11:11:33.000000000","tz":120},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I96fd8e6bec846065c9dc73d2f3af5899099461a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f6dbd116fd0891821cf37647eb7e10b490164328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/f6dbd116fd0891821cf37647eb7e10b490164328"}]},"branch":"refs/heads/master"},"11ff5ea21b805eae7314394ed657e4cab5dcf5c9":{"kind":"REWORK","_number":4,"created":"2024-09-30 07:54:06.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/51/926651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/51/926651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/4"}}},"commit":{"parents":[{"commit":"4e596cd865c0e23f98fab47d09e1cb31ed51b14e","subject":"Add release note for format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/4e596cd865c0e23f98fab47d09e1cb31ed51b14e"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-09-30 07:53:55.000000000","tz":120},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I96fd8e6bec846065c9dc73d2f3af5899099461a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/11ff5ea21b805eae7314394ed657e4cab5dcf5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/11ff5ea21b805eae7314394ed657e4cab5dcf5c9"}]},"branch":"refs/heads/master"},"fa2ecb269a4aaccfbb4903d95e8f13042d83aece":{"kind":"REWORK","_number":5,"created":"2024-09-30 08:25:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/926651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/51/926651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/51/926651/5"}}},"commit":{"parents":[{"commit":"d37a161b7c65a73a6aaabcb743b117cee41a0d58","subject":"Merge \"Add a function to trim scope out of ipv6 addr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/d37a161b7c65a73a6aaabcb743b117cee41a0d58"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 14:23:42.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-30 08:25:42.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I96fd8e6bec846065c9dc73d2f3af5899099461a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/fa2ecb269a4aaccfbb4903d95e8f13042d83aece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/fa2ecb269a4aaccfbb4903d95e8f13042d83aece"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
