)]}'
{"id":"openstack%2Foslo.utils~936273","triplet_id":"openstack%2Foslo.utils~master~Ia7ec96ace334bec4e6dc09f593f4052ec8e1de4e","project":"openstack/oslo.utils","branch":"master","topic":"bug/2089688","attention_set":{},"removed_from_attention_set":{"24434":{"account":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"last_update":"2024-11-27 15:11:01.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-27 13:12:46.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia7ec96ace334bec4e6dc09f593f4052ec8e1de4e","subject":"Fix get_my_ipv4 in absence of IPv4","status":"MERGED","created":"2024-11-26 15:42:40.000000000","updated":"2024-11-27 15:12:01.000000000","submitted":"2024-11-27 15:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936273-bug/2089688","meta_rev_id":"35ac9ebb2c636150a86cc2f50c91143f747c713c","_number":936273,"virtual_id_number":936273,"owner":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 15:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-27 13:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-27 13:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-11-27 13:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 15:52:08.000000000","updated_by":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-26 17:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-27 13:20:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"58bf13fa2706b0f5c8ddc56606479c3f95359a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-11-26 15:42:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3501eeb72808d85b2ad2b32f21f4e593b3f59da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-11-26 15:45:14.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5db8537856e4194cd04199ecf10ab557ba058aa7","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2024-11-26 15:52:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfd4d59bb2aa8a77dd1f6d443540068a5d6a3986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 17:10:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1015c89f91784446af2b1c6367b8afeb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0e2fdfffa0348bfac0a7f691d643db5 : SUCCESS in 1h 24m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fa197dae1284605bb0338e6ca728ed2 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d0858e4ab64ce9bc1a9a87d452e25d : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6fd92b69db84535ab3814ba57b17a3e : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40c9cc6b164a40e595af85171ed8adf4 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964b33a5207f464cb62b44ca7a3ec9e7 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"099943d4e2545c98671f8f7e452a4f4ed176ca68","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 13:12:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"002b540c971c36f9a7ada3865edb370ec6f88cbf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-27 13:20:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"359bfae0b07c2140961d3bccab8da3ae824f7027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 13:20:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"388ca0aa481578e3f5e3b03f8bb4ffff0f61f0b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 15:11:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4b3d90bbb04b60bc1097607c217796\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3a5857baad14d4d90817902b8fb1907 : SUCCESS in 1h 40m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e74739fbb88485a9ad07b19e2d4874f : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/326ea7cfc2054b41810ce5b764388a5a : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f8a08f9b76241e194c4c267be4b425d : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02fb942b0024d3797397b94971a531a : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d02941dd444cf2ccb541297c8c9692684717a2a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 15:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35ac9ebb2c636150a86cc2f50c91143f747c713c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 15:12:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd83aac6ef58400b87ef2ca5250d2021\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbf00d0edea40dd9c8b3802b2bab399 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b84a853663f4bcc05f83331aa3e5096d1c3d4ce","revisions":{"6ecd6b5f9d3b0690bf791dc2722cd211c30e2e8f":{"kind":"REWORK","_number":1,"created":"2024-11-26 15:42:40.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/73/936273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/73/936273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/1"}}},"commit":{"parents":[{"commit":"901916f08c28d2cb8585c6ba7666049da6dfaa9c","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/901916f08c28d2cb8585c6ba7666049da6dfaa9c"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-11-26 11:28:39.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-11-26 15:40:12.000000000","tz":60},"subject":"Fix get_my_ipv4 in absence of IPv4","message":"Fix get_my_ipv4 in absence of IPv4\n\nThe code assumes the existence of the IPv4 route-table,\nbut if IPv4 has been completely disabled, there is none.\n\nHandle it the same way as all the other errors and return\nthe link local IPv6 address.\n\nCloses-Bug: #2089688\nRelated-Bug: #2089670\nChange-Id: Ia7ec96ace334bec4e6dc09f593f4052ec8e1de4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/6ecd6b5f9d3b0690bf791dc2722cd211c30e2e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/6ecd6b5f9d3b0690bf791dc2722cd211c30e2e8f"}]},"branch":"refs/heads/master"},"8b84a853663f4bcc05f83331aa3e5096d1c3d4ce":{"kind":"NO_CHANGE","_number":2,"created":"2024-11-26 15:45:14.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/73/936273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/73/936273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/73/936273/2"}}},"commit":{"parents":[{"commit":"901916f08c28d2cb8585c6ba7666049da6dfaa9c","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/901916f08c28d2cb8585c6ba7666049da6dfaa9c"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-11-26 11:28:39.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2024-11-26 15:45:04.000000000","tz":60},"subject":"Fix get_my_ipv4 in absence of IPv4","message":"Fix get_my_ipv4 in absence of IPv4\n\nThe code assumes the existence of the IPv4 route-table,\nbut if IPv4 has been completely disabled, there is none.\n\nHandle it the same way as all the other errors and return\nthe link local IPv6 address.\n\nCloses-Bug: #2089688\nRelated-Bug: #2089670\nChange-Id: Ia7ec96ace334bec4e6dc09f593f4052ec8e1de4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/8b84a853663f4bcc05f83331aa3e5096d1c3d4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/8b84a853663f4bcc05f83331aa3e5096d1c3d4ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
