)]}'
{"id":"openstack%2Foslo.utils~944917","triplet_id":"openstack%2Foslo.utils~master~I3c6205ca518dc9c0fd35680b5559f9fd75f5879a","project":"openstack/oslo.utils","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-01 10:59:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c6205ca518dc9c0fd35680b5559f9fd75f5879a","subject":"Remove constant_time_prepare","status":"MERGED","created":"2025-03-18 17:31:33.000000000","updated":"2025-04-01 11:00:22.000000000","submitted":"2025-04-01 10:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944917","meta_rev_id":"de7107d43a51ea3ce9e5f5c62a9428ce36c1fd62","_number":944917,"virtual_id_number":944917,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 10:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-03-19 11:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-04-01 09:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-18 19:15:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-19 11:04:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"83ddfac81c0459fcc47a9da4ab33850b9df0e219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-18 17:31:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42c51564b6f315a18ca09e99dd6e4de76288e2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 19:15:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2562d0c5df0549c0a15a2449d1e28bf3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d5f2f05eba641eda58f4a2cd68250ec : SUCCESS in 1h 38m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63463a6436484560804a41e0b25a7db8 : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792142aa20c54b81a062c3f13439367b : FAILURE in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc9688530a344644a69762f068c2ec1e : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7058beb1589a47679fca048c254ee853 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb13cc53b43941a5b7ba0af4d97588c6 : FAILURE in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b62e7eb18bf64bc08872e08b65ff78c3 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1a01c4ea881866a0349250d304144708e0e0a6b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-19 01:11:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03974ff48575319535f244407b7740024b5b154c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 03:03:26.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/002682b7cc21434680d65eda52b70d2b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2410c79724f487891e5361467b86b88 : SUCCESS in 1h 47m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a41c0d0ca03740ba9f7ce277d7df7777 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8d870bc9074bcf9c33400d8022349d : FAILURE in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8c22d9ac59444d2aaa0c72cef13c99a : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6be45d7afcda4d3ba246217e5be7a203 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f504d33ecbf341c6b127da9629b8124a : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2afe47c975f2405e8f6ca7f185bd60a9 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1d4b59ff14f92209b33ad3724e8e702bb33841da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-19 06:54:07.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"387d077f2b63ca8939493a77465c7ea569ff9b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 08:43:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6212caa4a76148daa0f4f206c2ff361c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17eebf65ca36449087d503d68197962c : SUCCESS in 1h 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1e6637083924d78a3f52709a47182fb : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65866ba6f9a54a6cb80f88f6d291be6b : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58afdaaf10934a2a95bf6c69f3e1ea3b : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03643b654b2b4dd6b07844f7d6393226 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32aa7b6f1e3441918d790ba0af6c821b : SUCCESS in 2m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39af18e649154dd38572d73bf404d9c8 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d9c7dc8780c376b2852fccfecc9009796628d9d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-03-19 11:04:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f22c83d4f028fd21812095a95ced069b491c906e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-01 09:19:27.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b80c922c651b9fcdcd9ae4cd83421778f2f114f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:20:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f5eab887af43a20d6e2299d33a1b68e0bae8fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:59:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe6fa2160604021a395d98b164a7135\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b046af62a73470bb23641552f92ac2c : SUCCESS in 1h 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0c5b91f9c94c1cade317e4971ae38f : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecd16d8c59584ab383522c34cc0c9a20 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8aadd016bc947bda36740d408ad918b : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e53ea691d6b411ca6210d47bbcb46b1 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c6f6a0d12b2479c95369e443bb22d6a : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"31ad5ecae177a29aea143359c60b50f98e6422b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:59:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"de7107d43a51ea3ce9e5f5c62a9428ce36c1fd62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 11:00:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a501e4ba0aca4c2a9eca4459c5b9e526\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db7b8e361df489988ee616ea6aeb7a0 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c56c725a3cd4fe4bfcc37440c85fd3d : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f36253cb0cc5b81cfcd8cdf4144caa3beb22d33","revisions":{"a0f5028faeb7580a55d255e7204459e9f54ae243":{"kind":"REWORK","_number":1,"created":"2025-03-18 17:31:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/944917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/17/944917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/1"}}},"commit":{"parents":[{"commit":"dfdd8fa5575d23e1af42cca14d98038fe079d5d5","subject":"Merge \"Fix typo in recent change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/dfdd8fa5575d23e1af42cca14d98038fe079d5d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-18 17:30:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-18 17:30:41.000000000","tz":540},"subject":"Remove constant_time_prepare","message":"Remove constant_time_prepare\n\nWe no longer see any usage of the function in openstack namespace so\ncan remove it now.\n\nRelated-Bug: #2081732\nChange-Id: I3c6205ca518dc9c0fd35680b5559f9fd75f5879a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a0f5028faeb7580a55d255e7204459e9f54ae243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/a0f5028faeb7580a55d255e7204459e9f54ae243"}]},"branch":"refs/heads/master"},"6ae41202837b5cffe6446d0b449a08b638394339":{"kind":"REWORK","_number":2,"created":"2025-03-19 01:11:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/944917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/17/944917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/2"}}},"commit":{"parents":[{"commit":"67345b82e0368d19ae96c81c45d0a9683fa544ac","subject":"Fix undocumented functions in secretutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/67345b82e0368d19ae96c81c45d0a9683fa544ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-18 17:30:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-19 01:11:04.000000000","tz":540},"subject":"Remove constant_time_prepare","message":"Remove constant_time_prepare\n\nWe no longer see any usage of the function in openstack namespace so\ncan remove it now.\n\nRelated-Bug: #2081732\nChange-Id: I3c6205ca518dc9c0fd35680b5559f9fd75f5879a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/6ae41202837b5cffe6446d0b449a08b638394339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/6ae41202837b5cffe6446d0b449a08b638394339"}]},"branch":"refs/heads/master"},"2f36253cb0cc5b81cfcd8cdf4144caa3beb22d33":{"kind":"REWORK","_number":3,"created":"2025-03-19 06:54:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/944917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.utils","ref":"refs/changes/17/944917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.utils refs/changes/17/944917/3"}}},"commit":{"parents":[{"commit":"67345b82e0368d19ae96c81c45d0a9683fa544ac","subject":"Fix undocumented functions in secretutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/67345b82e0368d19ae96c81c45d0a9683fa544ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-18 17:30:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-19 06:54:07.000000000","tz":0},"subject":"Remove constant_time_prepare","message":"Remove constant_time_prepare\n\nWe no longer see any usage of the function in openstack namespace so\ncan remove it now.\n\nRelated-Bug: #2081732\nChange-Id: I3c6205ca518dc9c0fd35680b5559f9fd75f5879a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/2f36253cb0cc5b81cfcd8cdf4144caa3beb22d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.utils/commit/2f36253cb0cc5b81cfcd8cdf4144caa3beb22d33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
