)]}'
{"id":"openstack%2Foslo.versionedobjects~1003515","triplet_id":"openstack%2Foslo.versionedobjects~master~Icbaa829479a3d055d6eb7d734b3af8ba52654861","project":"openstack/oslo.versionedobjects","branch":"master","topic":"annotation-style-ovos","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-04 10:08:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Icbaa829479a3d055d6eb7d734b3af8ba52654861","subject":"Enable RUF rules","status":"MERGED","created":"2026-09-02 11:45:31.000000000","updated":"2026-09-04 11:44:50.000000000","submitted":"2026-09-04 11:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003515-annotation-style-ovos","meta_rev_id":"3102496554ea3c25a5f0a507b8ed14e78fd32af5","_number":1003515,"virtual_id_number":1003515,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 11:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-04 10:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-02 14:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-04 10:08:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 13:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 14:58:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1ba78e464e02c53b70c6bfaf7ee45e9757a50f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-02 11:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f0fabdbbad3477e20c83981fbdd8b021983a4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:07:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d440189b5d403e94feafbf801634a9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec4d20cb72c249129b4bfb3401c08027 : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0453c88c4de74f40af80699d78f6f524 : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f89450fe7f5469d849450ebeb5cd6bd : SUCCESS in 49m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624851016c2748389040e82afb1485c9 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23bc236675da45ccb9f8012e587d9bb6 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cfe4f68d85a4d6aaf7c857f3ee0a155 : SUCCESS in 5m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8042b7d0de24a7b8468c733854f107d : SUCCESS in 4m 00s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/df28a9e898464c758e9e3d9c614b7ee1 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c267b25dd4dc459f971fb369c3182a04 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62c9d107e91b4221b968e1a4fa57b78b : SUCCESS in 5m 32s\n- oslo.versionedobjects-src-grenade-multinode https://zuul.opendev.org/t/openstack/build/330b71fc49f245d1bf36adae1d16d3c6 : SUCCESS in 1h 15m 00s (non-voting)\n- oslo.versionedobjects-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/c97beb241a364c00a20cb621a074862c : SUCCESS in 13m 35s\n- oslo.versionedobjects-ironic-tox-py313 https://zuul.opendev.org/t/openstack/build/99d55b9efdf84a219e58770b5fc6c117 : SUCCESS in 8m 09s\n\nWarning:\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/basic_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/compound_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/enum_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/object_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/subclassing.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/timestamp_mixin.py","accounts_in_message":[],"_revision_number":1},{"id":"781192730000136548b5c5cfcface65a5d833f01","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-02 14:58:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d083ec67c7a745db7775a9438c8760fb71455081","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 10:08:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"904aa74de41385e9e37c14c5f8117f966ab02673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 10:08:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25fb971fb7a3ca2e7f5d5211b35ebdebd7a1c5ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:43:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99222bdcebd4c7fa22721cb57531ae3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e96051e163d424cb2cf975269fc75b0 : SUCCESS in 3m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6917a02e0eb4f9abf48827869941782 : SUCCESS in 1m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/294365ff06df48f99ca3eca7df76e037 : SUCCESS in 1h 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a435390da247669a495ea931295e6d : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ccac1c47d82439db8dee8b17760afc5 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22a56d817bda4150865658c4cda3e5d6 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96dcd9a800f54b3eacb03a60dd71dd2a : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d60a1632634561a2534b399a06111c : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b899481f34b4955a184c35d82c4d891 : SUCCESS in 5m 44s\n- oslo.versionedobjects-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/9cd7b175e20148bba3eb7429f6893262 : SUCCESS in 9m 39s\n- oslo.versionedobjects-ironic-tox-py313 https://zuul.opendev.org/t/openstack/build/f78dcdcd71e24435beac98ca9a116f89 : SUCCESS in 7m 28s\n\nWarning:\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/basic_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/compound_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/enum_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/object_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/subclassing.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/timestamp_mixin.py","accounts_in_message":[],"_revision_number":1},{"id":"b33a26fa696b957b9c6e10dd94ef4c20d920fc88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:43:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3102496554ea3c25a5f0a507b8ed14e78fd32af5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 11:44:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4f2526b2994223bc5d1fe04f0c4adf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa2a203ab4243d08c638fca631f28b8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c50ddc9602f3b249d4117af2d3ebdd5e1ec4e7d","revisions":{"9c50ddc9602f3b249d4117af2d3ebdd5e1ec4e7d":{"kind":"REWORK","_number":1,"created":"2026-09-02 11:45:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/1003515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.versionedobjects","ref":"refs/changes/15/1003515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/15/1003515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/15/1003515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/15/1003515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/15/1003515/1"}}},"commit":{"parents":[{"commit":"6fc2dcc0dfc3df98e94131dd7d06320cebb97f3d","subject":"Merge \"Deprecate item_cls argument to obj_make_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/6fc2dcc0dfc3df98e94131dd7d06320cebb97f3d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-12 11:31:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-31 22:05:07.000000000","tz":60},"subject":"Enable RUF rules","message":"Enable RUF rules\n\nThis is part of the new defaults in ruff 0.16.x, which we bump to here.\nWe disable RUF012 since we don\u0027t want to annotate all \u0027fields\u0027 instances\nwith \u0027ClassVar\u0027, while the other things it catches are auto-resolved. We\nalso have to disable two instances of RUF005, since applying the fixes\nwould change the semantics of the test.\n\nChange-Id: Icbaa829479a3d055d6eb7d734b3af8ba52654861\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/9c50ddc9602f3b249d4117af2d3ebdd5e1ec4e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/9c50ddc9602f3b249d4117af2d3ebdd5e1ec4e7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
