)]}'
{"id":"openstack%2Foslo.versionedobjects~1004446","triplet_id":"openstack%2Foslo.versionedobjects~master~Iccd9b5b1a9045936b221c52f6365f07c83130025","project":"openstack/oslo.versionedobjects","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 21:57:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 23:20:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iccd9b5b1a9045936b221c52f6365f07c83130025","subject":"Resolve setuptools deprecation warnings","status":"MERGED","created":"2026-09-07 16:00:14.000000000","updated":"2026-09-07 23:29:58.000000000","submitted":"2026-09-07 23:27:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004446","meta_rev_id":"6c353faddbb41d98e69c6bd9440150d02674d576","_number":1004446,"virtual_id_number":1004446,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 21:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 21:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 17:30:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 18:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"083f4883204ca1dc18bf2711be84a9ac638ebd9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 16:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a81d38cfa377393056075122426517b2220b9d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 17:30:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e8176052369857be99db25d73b5cdea7621441a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:33:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3c3c35ec2193ca5ff5f44cef824905d68f9fe5c7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 17:33:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8f1c94f3b3903174a1123c7255fdb3786fd6784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 18:57:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2beb5b10de6a4bb7a9475eefa8f9dd25\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2d8ee84db6445f7844b156d293cdd89 : SUCCESS in 4m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3389a8cf5e6042a9b83c3b25a04739e1 : SUCCESS in 3m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff2d00b076b0448398d0edbdad461a2f : SUCCESS in 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02906984beb454c8bde6af8c7113d59 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0977ff4960e24c4d93d23eb47070dccb : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db93886568dc4fd7a016cc56cdcaae15 : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d22c044219f74c258231253257b24e0c : SUCCESS in 6m 05s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3d315274cbd3455b985a196724bc777b : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfa98f3eed048b8bace7ce1719d9bcb : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2471aedf43eb4481a088814eaccfde52 : SUCCESS in 5m 54s\n- oslo.versionedobjects-src-grenade-multinode https://zuul.opendev.org/t/openstack/build/eba8e1d5ce024e8f98eeb808d784297b : SUCCESS in 1h 18m 45s (non-voting)\n- oslo.versionedobjects-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/b2c610785b2f43fcb7600279daef0092 : SUCCESS in 8m 11s\n- oslo.versionedobjects-ironic-tox-py313 https://zuul.opendev.org/t/openstack/build/8787a053de074d0aa281f585f6762912 : SUCCESS in 6m 33s\n\nWarning:\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/basic_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/compound_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/enum_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/object_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/subclassing.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/timestamp_mixin.py","accounts_in_message":[],"_revision_number":2},{"id":"024b138fde6d12451d834e57ca74534122308b2f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 21:57:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa580436768ff0658dd4eda49ce268295dbd1906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:58:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81ab15297a257db65ce4c2522c5dcd06d3bb254e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 23:20:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2214678818a5eb01b226e9e21b0275ca7423146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:27:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1263a50f5041c787ef47e4b715aac1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e731dc083d4c4c5aa2e5228e8cbd1550 : SUCCESS in 4m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/021d780a39834481a2d82bfc0119261d : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb85862536d45bb85633589cb806c9b : SUCCESS in 1h 23m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73d9da25b4947d0aeb6e26ef55a54e5 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0136e02cd0bd482691f2b64a3431809a : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29ad285098d6404088eafa2c517da128 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a9c1f1239dd43f081ee79026aa169b8 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53df1a60863e458e867796f91d97cbe5 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0981f2585f74d1baa37374a7661b2e5 : SUCCESS in 5m 35s\n- oslo.versionedobjects-nova-tox-py313 https://zuul.opendev.org/t/openstack/build/d9e94dd6100143db999a1ac980834589 : SUCCESS in 14m 02s\n- oslo.versionedobjects-ironic-tox-py313 https://zuul.opendev.org/t/openstack/build/1da3a2240a5b46468a87179554cd01c1 : SUCCESS in 7m 06s\n\nWarning:\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/basic_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/compound_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/enum_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/object_fields.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/subclassing.py\n  Comments left for invalid file oslo_versionedobjects/tests/test_data/timestamp_mixin.py","accounts_in_message":[],"_revision_number":2},{"id":"05b30dfe26cd15e6905c48188df4bdf5bf054e60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:27:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6c353faddbb41d98e69c6bd9440150d02674d576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:29:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc0311550b84ad29efd93d54771bc96\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3421044bddcc4a079b280733d2a3612b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b72957c8c8ca9057769483549669182b4b4515a3","revisions":{"915d91a1c26f7c6598d376a8535ca44aa21ecd36":{"kind":"REWORK","_number":1,"created":"2026-09-07 16:00:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1004446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.versionedobjects","ref":"refs/changes/46/1004446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/1"}}},"commit":{"parents":[{"commit":"1f06585282a986cdee5c915bb469e0e5f113c0e7","subject":"Merge \"Enable RUF rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/1f06585282a986cdee5c915bb469e0e5f113c0e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:59:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:59:58.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: Iccd9b5b1a9045936b221c52f6365f07c83130025\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/915d91a1c26f7c6598d376a8535ca44aa21ecd36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/915d91a1c26f7c6598d376a8535ca44aa21ecd36"}]},"branch":"refs/heads/master"},"b72957c8c8ca9057769483549669182b4b4515a3":{"kind":"REWORK","_number":2,"created":"2026-09-07 17:33:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/1004446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.versionedobjects","ref":"refs/changes/46/1004446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/46/1004446/2"}}},"commit":{"parents":[{"commit":"1f06585282a986cdee5c915bb469e0e5f113c0e7","subject":"Merge \"Enable RUF rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/1f06585282a986cdee5c915bb469e0e5f113c0e7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:59:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 17:33:35.000000000","tz":540},"subject":"Resolve setuptools deprecation warnings","message":"Resolve setuptools deprecation warnings\n\nChange-Id: Iccd9b5b1a9045936b221c52f6365f07c83130025\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/b72957c8c8ca9057769483549669182b4b4515a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/b72957c8c8ca9057769483549669182b4b4515a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
