)]}'
{"id":"openstack%2Foslo.versionedobjects~722733","triplet_id":"openstack%2Foslo.versionedobjects~master~I9fb0d3558e4aa2576011f05011dbc8623c77c6b8","project":"openstack/oslo.versionedobjects","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"I9fb0d3558e4aa2576011f05011dbc8623c77c6b8","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:25:59.000000000","updated":"2020-04-27 10:07:43.000000000","submitted":"2020-04-27 10:06:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722733-1587981965399-16b58591","meta_rev_id":"48dcb23fbe7e1e9ee574b0e42498e7d8f320a15a","_number":722733,"virtual_id_number":722733,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 10:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-27 08:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-24 16:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-27 08:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 16:47:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-27 08:26:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-27 10:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"943f4865810c993125f26ec3e046a315dff1fe78","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:25:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f3184355b972d1f46f1918fdd674e051c8fec8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:53:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8d3b8757aa7e4205ac5ed0259206aa89 : SUCCESS in 4m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f58ad52e886a4fa5b741cb5072d7f7ce : SUCCESS in 1h 18m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f10e4497be3c40829b4f51df38763dae : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ab0853627445139fc2cee987f501de : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b4cd1a6c6b47b0804de8391d8dd623 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3f433e1e8114873bf895f491ecebc6a : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed18568807247bd85bd2c932d387030 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"07032c906a44f469c1b77801a19b4932c28db042","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-24 16:47:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"679580515c709a0f05741f9cc3db1244c7685c50","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-27 08:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f7c5dad910c122ff169b055f062cf8e08baabba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:27:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0e3a7f89da2127e31f556d1700d886f371d6ba8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:06:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52fde4f7730e4de8855df4d15f8e658d : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d9b84c8e2a3434cb09a80678d0bfade : SUCCESS in 1h 36m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7042591941fc4b0792b7164ad94fecac : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b20a3fc6674d55a788e14faef9ab4d : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302cca65472c4e318d50271c101865f9 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa1c4c121fc4d14a9b4259bf74a4b59 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c415ccc9c143c2911b01cf4dff571e : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"49b017be18369c02c219777e750a9f0113544059","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:06:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4679ab1e3bbd3ec957233c3317ecf9f20f23043a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:07:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c35ed361bff47f0b553ed3abedabd7e : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b9f28411698fcb995013e4618dd192791d6ca04","revisions":{"6b9f28411698fcb995013e4618dd192791d6ca04":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:25:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/33/722733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.versionedobjects","ref":"refs/changes/33/722733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/33/722733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/33/722733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/33/722733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.versionedobjects refs/changes/33/722733/1"}}},"commit":{"parents":[{"commit":"5655e7f7b2f032edbca776dd768f0195305799ff","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/5655e7f7b2f032edbca776dd768f0195305799ff"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:14.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:23:14.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: I9fb0d3558e4aa2576011f05011dbc8623c77c6b8\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/6b9f28411698fcb995013e4618dd192791d6ca04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.versionedobjects/commit/6b9f28411698fcb995013e4618dd192791d6ca04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
