)]}'
{"id":"openstack%2Foslo.vmware~343379","triplet_id":"openstack%2Foslo.vmware~master~I7a4f4a83a394585cdce7a4b2ebbaced7aa034dbd","project":"openstack/oslo.vmware","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I7a4f4a83a394585cdce7a4b2ebbaced7aa034dbd","subject":"Updated from global requirements","status":"MERGED","created":"2016-07-17 23:59:00.000000000","updated":"2016-07-18 11:12:31.000000000","submitted":"2016-07-18 11:12:31.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1fed8f6ba5ee6acbc1f388e80cd3b4fe1c25e1fc","_number":343379,"virtual_id_number":343379,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":2,"date":"2016-07-18 11:12:31.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"all":[{"value":2,"date":"2016-07-18 10:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"all":[{"value":1,"date":"2016-07-18 10:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-18 10:09:22.000000000","updated_by":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"reviewer":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"state":"REVIEWER"},{"updated":"2016-07-18 11:12:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7ab85c32f027985a78b4f52e7ed35288149a8ae5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-07-17 23:59:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c55469705f7a5d60b97246bc579ef9bbff40411e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-18 01:04:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.vmware-python34 http://logs.openstack.org/79/343379/1/check/gate-oslo.vmware-python34/4850f7b/ : SUCCESS in 2m 02s\n- gate-oslo.vmware-python35 http://logs.openstack.org/79/343379/1/check/gate-oslo.vmware-python35/80d194f/ : SUCCESS in 7m 38s\n- gate-oslo.vmware-docs http://docs-draft.openstack.org/79/343379/1/check/gate-oslo.vmware-docs/26dd9e1//doc/build/html/ : SUCCESS in 2m 06s\n- gate-oslo.vmware-pep8 http://logs.openstack.org/79/343379/1/check/gate-oslo.vmware-pep8/6ce4a1c/ : SUCCESS in 5m 12s\n- gate-oslo.vmware-python27 http://logs.openstack.org/79/343379/1/check/gate-oslo.vmware-python27/2091829/ : SUCCESS in 4m 46s\n- gate-oslo.vmware-requirements http://logs.openstack.org/79/343379/1/check/gate-oslo.vmware-requirements/00d751e/ : SUCCESS in 1m 49s\n- gate-tempest-dsvm-neutron-src-oslo.vmware http://logs.openstack.org/79/343379/1/check/gate-tempest-dsvm-neutron-src-oslo.vmware/22ebd63/ : SUCCESS in 56m 43s\n- gate-tempest-dsvm-neutron-src-oslo.vmware-mitaka http://logs.openstack.org/79/343379/1/check/gate-tempest-dsvm-neutron-src-oslo.vmware-mitaka/a89f378/ : SUCCESS in 53m 46s (non-voting)\n- gate-oslo.vmware-releasenotes http://docs-draft.openstack.org/79/343379/1/check/gate-oslo.vmware-releasenotes/f31a95b//releasenotes/build/html/ : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e70c8c53e83b29bd88b48f4dc5f4da2641fdc814","author":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"date":"2016-07-18 10:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6840b0f2e8379cd1fbee5e1ac6cbda18e67bc69f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-18 10:13:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"835a4cdd47da8019cc7cc1acb59dec31b0b236e9","date":"2016-07-18 11:12:31.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"b58ee9e63a4de7d9df185ec7aae31e1bad58bc67","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-18 11:12:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.vmware-python34 http://logs.openstack.org/79/343379/1/gate/gate-oslo.vmware-python34/8506fbf/ : SUCCESS in 2m 30s\n- gate-oslo.vmware-python35 http://logs.openstack.org/79/343379/1/gate/gate-oslo.vmware-python35/e45922b/ : SUCCESS in 3m 05s\n- gate-oslo.vmware-docs http://docs-draft.openstack.org/79/343379/1/gate/gate-oslo.vmware-docs/c540d83//doc/build/html/ : SUCCESS in 3m 44s\n- gate-oslo.vmware-pep8 http://logs.openstack.org/79/343379/1/gate/gate-oslo.vmware-pep8/bd51e04/ : SUCCESS in 3m 30s\n- gate-oslo.vmware-python27 http://logs.openstack.org/79/343379/1/gate/gate-oslo.vmware-python27/2a8468c/ : SUCCESS in 2m 57s\n- gate-oslo.vmware-requirements http://logs.openstack.org/79/343379/1/gate/gate-oslo.vmware-requirements/beb0720/ : SUCCESS in 1m 53s\n- gate-tempest-dsvm-neutron-src-oslo.vmware http://logs.openstack.org/79/343379/1/gate/gate-tempest-dsvm-neutron-src-oslo.vmware/bea58f1/ : SUCCESS in 58m 34s\n- gate-oslo.vmware-releasenotes http://docs-draft.openstack.org/79/343379/1/gate/gate-oslo.vmware-releasenotes/84e543f//releasenotes/build/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"170d6b7b82826a6314e158d4ecc3a4414e921a10","revisions":{"170d6b7b82826a6314e158d4ecc3a4414e921a10":{"kind":"REWORK","_number":1,"created":"2016-07-17 23:59:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/343379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/79/343379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/343379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/343379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/343379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/79/343379/1"}}},"commit":{"parents":[{"commit":"421b41c15b76781cc34bad16cd242a49517529ce","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/421b41c15b76781cc34bad16cd242a49517529ce"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-07-17 23:58:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-07-17 23:58:59.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I7a4f4a83a394585cdce7a4b2ebbaced7aa034dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/170d6b7b82826a6314e158d4ecc3a4414e921a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/170d6b7b82826a6314e158d4ecc3a4414e921a10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"421b41c15b76781cc34bad16cd242a49517529ce","is_merged_in_target_branch":true,"change_id":"I5cde277361a1bcd2b222ad9aebabc24dd0fbe977","change_number":340044,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
