)]}'
{"id":"openstack%2Foslo.vmware~658513","triplet_id":"openstack%2Foslo.vmware~master~I36feb93bc23ba5ef0cb497859140bb5f962f2bee","project":"openstack/oslo.vmware","branch":"master","topic":"bandit","hashtags":[],"change_id":"I36feb93bc23ba5ef0cb497859140bb5f962f2bee","subject":"Add nosec to skip bandit check","status":"MERGED","created":"2019-05-11 01:29:57.000000000","updated":"2019-05-14 10:19:14.000000000","submitted":"2019-05-14 10:11:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658513-1557828714690-96febb6b","meta_rev_id":"a72f59e883f98ffe527f4765a9c8adc395fb83b0","_number":658513,"virtual_id_number":658513,"owner":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-11 01:35:37.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-13 18:45:33.000000000","_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":2,"date":"2019-05-14 10:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-13 18:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":2,"date":"2019-05-13 18:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":1,"date":"2019-05-14 08:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 01:35:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 18:36:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-14 08:31:27.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2019-05-14 10:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c26658763219e8cd8bbc8411bd511f144e010916","author":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"date":"2019-05-11 01:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dfa4e8fad711f68b537d786a1786973cd869e1b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-11 01:35:37.000000000","message":"Patch Set 1:\n\n- ext-oslo.vmware-dsvm-nsx http://207.189.188.190/logs/ext-oslo.vmware-dsvm-nsx/658513/1/193/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"ac5a30cc40afa5f9494a16e9780f19cc8e47ffbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 03:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/13/658513/1/check/tempest-full/25de085/ : SUCCESS in 1h 44m 39s\n- tempest-full-py3 http://logs.openstack.org/13/658513/1/check/tempest-full-py3/f224689/ : SUCCESS in 1h 30m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/658513/1/check/openstack-tox-lower-constraints/8bcecac/ : SUCCESS in 8m 08s\n- openstack-tox-pep8 http://logs.openstack.org/13/658513/1/check/openstack-tox-pep8/a7398ac/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/13/658513/1/check/openstack-tox-py27/afe510b/ : SUCCESS in 7m 31s\n- openstack-tox-py35 http://logs.openstack.org/13/658513/1/check/openstack-tox-py35/6c7e0a6/ : SUCCESS in 6m 39s\n- openstack-tox-py36 http://logs.openstack.org/13/658513/1/check/openstack-tox-py36/e81c327/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/13/658513/1/check/openstack-tox-py37/ab2cf11/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/13/658513/1/check/openstack-tox-docs/14c0d17/html/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7534a59ee091de605837fa49c6278f364aa120e7","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2019-05-13 18:36:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98a2bba77eb3da01cd9ee1f02c947515987d1875","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 18:36:45.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good. As an alternative, it looks like it would be possible to just use a string other than \u0027password\u0027 for this. Or maybe even better, update the bandit config to not run against the unit test code.","accounts_in_message":[],"_revision_number":1},{"id":"2e1362d501e6d058dd298040d441e97732bf319e","author":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"date":"2019-05-13 18:45:33.000000000","message":"Patch Set 1:\n\n@Sean I tried other string, but it is the variable name which is triggering that rule.\n\nSee https://github.com/PyCQA/bandit/blob/1.6.0/bandit/plugins/general_hardcoded_password.py#L52","accounts_in_message":[],"_revision_number":1},{"id":"0338b5d8651c64ad8a96760298a65520300ba469","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2019-05-14 08:31:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbe3cf80e9612b5c666bdf1d4f07e032debba29b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 08:31:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31e882951dfd0a26700c793544ab2407122328ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:11:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2cf154599ee1927c00372abd5ec9b95a2614815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:11:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/13/658513/1/gate/tempest-full/350da3f/ : SUCCESS in 1h 39m 54s\n- tempest-full-py3 http://logs.openstack.org/13/658513/1/gate/tempest-full-py3/0169892/ : SUCCESS in 1h 22m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/658513/1/gate/openstack-tox-lower-constraints/957c81b/ : SUCCESS in 6m 37s\n- openstack-tox-pep8 http://logs.openstack.org/13/658513/1/gate/openstack-tox-pep8/d1005a6/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/13/658513/1/gate/openstack-tox-py27/ab3a883/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/13/658513/1/gate/openstack-tox-py35/44bbbae/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/13/658513/1/gate/openstack-tox-py36/5acb13b/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/13/658513/1/gate/openstack-tox-py37/397bfbc/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/13/658513/1/gate/openstack-tox-docs/8508449/html/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cb1e20e2792f61dec07c7a3b0c784eb22fa2dad9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-14 10:19:14.000000000","message":"Patch Set 1:\n\nSee [1] and [2]. This is a bug with bandit. We shouldn\u0027t be checking unit test code ever\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006116.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006186.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95d849ebe733c3748bdc293362d3a9a4b744b857","revisions":{"95d849ebe733c3748bdc293362d3a9a4b744b857":{"kind":"REWORK","_number":1,"created":"2019-05-11 01:29:57.000000000","uploader":{"_account_id":9171,"name":"Vipin Balachandran","email":"vipin.bl@gmail.com","username":"vbala"},"ref":"refs/changes/13/658513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/13/658513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/13/658513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/13/658513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/13/658513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/13/658513/1"}}},"commit":{"parents":[{"commit":"dda60b5204a137508a6412d7f87ae01922f0bf06","subject":"Merge \"When looping call fails do not log the exception trace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/dda60b5204a137508a6412d7f87ae01922f0bf06"}]}],"author":{"name":"Vipin Balachandran","email":"vbala@vmware.com","date":"2019-05-11 01:27:04.000000000","tz":-420},"committer":{"name":"Vipin Balachandran","email":"vbala@vmware.com","date":"2019-05-11 01:27:04.000000000","tz":-420},"subject":"Add nosec to skip bandit check","message":"Add nosec to skip bandit check\n\nBandit is flagging one of the lines in unit test as a security\nissue which is false positive. Adding nosec to skip that line.\n\nChange-Id: I36feb93bc23ba5ef0cb497859140bb5f962f2bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/95d849ebe733c3748bdc293362d3a9a4b744b857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/95d849ebe733c3748bdc293362d3a9a4b744b857"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
