)]}'
{"id":"openstack%2Foslo.vmware~669464","triplet_id":"openstack%2Foslo.vmware~master~I4625a70e74f5c01c7b69ef14f1112c5057d10255","project":"openstack/oslo.vmware","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I4625a70e74f5c01c7b69ef14f1112c5057d10255","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:09:12.000000000","updated":"2019-07-19 21:18:31.000000000","submitted":"2019-07-19 21:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669464-1563571111731-6d7173bf","meta_rev_id":"7356a3d054e0082f6dddaab7ff1858070c7e195d","_number":669464,"virtual_id_number":669464,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-19 09:58:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-19 21:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 09:50:21.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-19 09:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-07-19 09:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 14:29:51.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-19 09:50:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-19 09:58:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-19 21:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"548fbbfca6ef63db879d8f75e26644f4b8b1ab4f","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:09:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa543f37e70825aeab6f4f4ce55ce362c470134d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-05 19:42:03.000000000","message":"Patch Set 1:\n\n- ext-oslo.vmware-dsvm-nsx http://207.189.188.190/logs/ext-oslo.vmware-dsvm-nsx/669464/1/207/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"d642c770bdf62cc46fc50292e1487b3b8a6cf923","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:04:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/64/669464/1/check/test-release-openstack/f4b14a1/ : SUCCESS in 2m 20s\n- tempest-full http://logs.openstack.org/64/669464/1/check/tempest-full/7526f47/ : SUCCESS in 1h 47m 18s\n- tempest-full-py3 http://logs.openstack.org/64/669464/1/check/tempest-full-py3/626ad59/ : SUCCESS in 1h 49m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/669464/1/check/openstack-tox-lower-constraints/93f1ccb/ : FAILURE in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/64/669464/1/check/openstack-tox-pep8/f1f2d00/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/64/669464/1/check/openstack-tox-py27/d11bc71/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/64/669464/1/check/openstack-tox-py36/ff653dd/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/64/669464/1/check/openstack-tox-py37/6624aec/ : SUCCESS in 3m 50s\n- openstack-tox-docs http://logs.openstack.org/64/669464/1/check/openstack-tox-docs/2d4b8d8/html/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/64/669464/1/check/build-openstack-releasenotes/cba0599/html/ : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"63dd72ce2e78c80682e0665160d79e375bb9ba41","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-08 14:29:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f5ec1130d8c57d0281c0ddc691956b051dc3e8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-19 09:50:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"724551769d17ea72e8aaf0952f6affafe9c48cc1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-19 09:50:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"65034e4ede29202982968eee34bfc20ad57b9ef2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 09:58:17.000000000","message":"Patch Set 2:\n\n- ext-oslo.vmware-dsvm-nsx http://207.189.188.190/logs/ext-oslo.vmware-dsvm-nsx/669464/2/212/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"5083a6d3121c2864e460e0b4cdd0d33961e6b1c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 11:37:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/64/669464/2/check/test-release-openstack/19c9123/ : SUCCESS in 2m 07s\n- tempest-full http://logs.openstack.org/64/669464/2/check/tempest-full/cd92730/ : SUCCESS in 1h 41m 53s\n- tempest-full-py3 http://logs.openstack.org/64/669464/2/check/tempest-full-py3/549c39c/ : SUCCESS in 1h 27m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/669464/2/check/openstack-tox-lower-constraints/8c2be6e/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 http://logs.openstack.org/64/669464/2/check/openstack-tox-pep8/4415a57/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/64/669464/2/check/openstack-tox-py27/33c68ee/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/64/669464/2/check/openstack-tox-py36/081c489/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/64/669464/2/check/openstack-tox-py37/1410f7d/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/64/669464/2/check/openstack-tox-docs/fef7208/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/64/669464/2/check/build-openstack-releasenotes/40c0da5/html/ : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7e65b4b06621954517fecf9ba93726ef2507e656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 19:22:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65c42aa6318bbfaeeceff5ed0b4b40eb33c95ebe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 21:18:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"006758dfcffc5175ca8cfccd387971fdda1d6f0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 21:18:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/64/669464/2/gate/test-release-openstack/822e7e9/ : SUCCESS in 1m 54s\n- tempest-full http://logs.openstack.org/64/669464/2/gate/tempest-full/1cac334/ : SUCCESS in 1h 42m 02s\n- tempest-full-py3 http://logs.openstack.org/64/669464/2/gate/tempest-full-py3/4f4b7dd/ : SUCCESS in 1h 25m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/64/669464/2/gate/openstack-tox-lower-constraints/e2f4e1c/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/64/669464/2/gate/openstack-tox-pep8/7d4a7f9/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/64/669464/2/gate/openstack-tox-py27/b465155/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/64/669464/2/gate/openstack-tox-py36/9d015a0/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/64/669464/2/gate/openstack-tox-py37/fa60e6b/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/64/669464/2/gate/openstack-tox-docs/1c01408/html/ : SUCCESS in 2m 51s\n- build-openstack-releasenotes http://logs.openstack.org/64/669464/2/gate/build-openstack-releasenotes/ecd991e/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1c25b189fc78e333ba2ace781b6e90cca853bde","revisions":{"c2254eeffe955d89dc8101091e909ec783a12eb6":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:09:12.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/64/669464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/64/669464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/1"}}},"commit":{"parents":[{"commit":"d63f76a9b25518f21ca6756fa9e3b85f2c9ac359","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d63f76a9b25518f21ca6756fa9e3b85f2c9ac359"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:58.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:09:11.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I4625a70e74f5c01c7b69ef14f1112c5057d10255\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/c2254eeffe955d89dc8101091e909ec783a12eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/c2254eeffe955d89dc8101091e909ec783a12eb6"}]},"branch":"refs/heads/master"},"f1c25b189fc78e333ba2ace781b6e90cca853bde":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-19 09:50:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/669464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/64/669464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/64/669464/2"}}},"commit":{"parents":[{"commit":"ce3122c2f17252ac2e82676cd675f62bab8c5420","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/ce3122c2f17252ac2e82676cd675f62bab8c5420"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:58.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-07-19 09:50:21.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I4625a70e74f5c01c7b69ef14f1112c5057d10255\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/f1c25b189fc78e333ba2ace781b6e90cca853bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/f1c25b189fc78e333ba2ace781b6e90cca853bde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
