)]}'
{"id":"openstack%2Foslo.vmware~710741","triplet_id":"openstack%2Foslo.vmware~master~I7b1fd3412b815d88b6d54178566f4089433f98ff","project":"openstack/oslo.vmware","branch":"master","topic":"drop-six","hashtags":[],"change_id":"I7b1fd3412b815d88b6d54178566f4089433f98ff","subject":"Drop use of six","status":"MERGED","created":"2020-03-02 11:13:05.000000000","updated":"2020-03-21 22:07:16.000000000","submitted":"2020-03-21 22:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710741-1584828359277-1d1e633c","meta_rev_id":"e13ff8b34dd271c25bb7d241680ba3e464e178c5","_number":710741,"virtual_id_number":710741,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 22:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-21 20:43:24.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-11 14:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-21 20:43:24.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 14:47:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-21 20:43:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-21 22:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16cab61290c290d71aab3ae89fefca0d94482b93","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 11:13:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf316216643787f44f56d2343589a4f5e824a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:36:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/592790b365d6486bb8edf5018cef05ac : SUCCESS in 4m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebda02f5cf0e449ebdd0a05a281da78a : SUCCESS in 1h 21m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b026644505754583bc7a8696ad3cdbde : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77de14e671cd4d94b6d2dd51808db338 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cb27175ab454f9c85ffed7f4838e15b : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eeccc12686c4fb78f4378ef1c03ef56 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67c528ac66644964a7b1517c0f7175c4 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f864787f4d4f32ae5b47007c2d76b2 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2c16260fe969fa09962acc0324be1995d5fd1e3d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 13:50:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11613b9362757e4ccc6b476f23d3ae89ce566649","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 13:52:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"077b4af627eb5a4a7e832585418d51e58cd0019c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 15:28:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f52ef4e79b69470dbee7e8630d4ad8c8 : SUCCESS in 3m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc05bce491d34533949d3b325f22aa48 : SUCCESS in 1h 28m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d840dd1573b64a569e2cad27346a1af7 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5149c58b7644748def10a6b293d5a4 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e354b78ed44fb58559dee011504126 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d55ee529b8e4356addd901e517b49ef : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b19f6ab840e4aefa93942f6c4b49392 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946a63bafd784c85b823af790021fbf0 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f69c3394aab1bcb9acdbd09a3eee9bd4730924ae","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-11 14:47:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nIdea for a follow-up there. Nothing block-worthy","accounts_in_message":[],"_revision_number":3},{"id":"194391f9058194c6b0c24e77f60bc02265351aca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-21 20:43:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nGood thing to follow up with a cleanup, but not worth blocking the rest of this.","accounts_in_message":[],"_revision_number":3},{"id":"314956a25e927d7e56897d6cceecd45df54a4920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 20:43:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9111e84779e89ef56294c3ba88c659b28debbc50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 22:05:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/df12d41654ee4fe48735cd7d7c2fed7b : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2b805679dd5499f86d8860c2cbacd73 : SUCCESS in 1h 17m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59d7af1b5cce415bb2d1011c021b983e : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36b5ee08300492d87d6d5d2be6837fa : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67735b9fc214482da006c14fb158ca50 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3634c32bab3e45c29acef753f8c242a7 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d159cbd65741f7955d139d3af45a6a : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"aa4e94dfe17a92a0cce4f3d9864ad83d720fd53c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 22:05:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"060e19819363a72aade360206aa521e791dcd726","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 22:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44fa6f2da3d4c86b84a155c397ac074 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88798e9ad8d251dfd76ff34be622c467b1b2d94a","revisions":{"72cb6393632deac6ff6d0384cf09736437a1388b":{"kind":"REWORK","_number":1,"created":"2020-03-02 11:13:05.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/710741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/41/710741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/1"}}},"commit":{"parents":[{"commit":"fbb6cedfbebda04613e8176eb109bfb3018486dc","subject":"Merge \"Adds standard file operation functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/fbb6cedfbebda04613e8176eb109bfb3018486dc"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 11:12:37.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 11:12:37.000000000","tz":60},"subject":"Drop use of six","message":"Drop use of six\n\nChange-Id: I7b1fd3412b815d88b6d54178566f4089433f98ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/72cb6393632deac6ff6d0384cf09736437a1388b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/72cb6393632deac6ff6d0384cf09736437a1388b"}]},"branch":"refs/heads/master"},"d0eca399f470018d9f09caacaec069403ef85af4":{"kind":"REWORK","_number":2,"created":"2020-03-02 13:50:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/710741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/41/710741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/2"}}},"commit":{"parents":[{"commit":"cb40fac97296d431a32585cf40ef5ef80e8e9075","subject":"Merge \"Add ability to change read chunk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/cb40fac97296d431a32585cf40ef5ef80e8e9075"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 11:12:37.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 13:50:43.000000000","tz":60},"subject":"Drop use of six","message":"Drop use of six\n\nChange-Id: I7b1fd3412b815d88b6d54178566f4089433f98ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d0eca399f470018d9f09caacaec069403ef85af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d0eca399f470018d9f09caacaec069403ef85af4"}]},"branch":"refs/heads/master"},"88798e9ad8d251dfd76ff34be622c467b1b2d94a":{"kind":"REWORK","_number":3,"created":"2020-03-02 13:52:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/710741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/41/710741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/41/710741/3"}}},"commit":{"parents":[{"commit":"cb40fac97296d431a32585cf40ef5ef80e8e9075","subject":"Merge \"Add ability to change read chunk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/cb40fac97296d431a32585cf40ef5ef80e8e9075"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 11:12:37.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-02 13:52:23.000000000","tz":60},"subject":"Drop use of six","message":"Drop use of six\n\nChange-Id: I7b1fd3412b815d88b6d54178566f4089433f98ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/88798e9ad8d251dfd76ff34be622c467b1b2d94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/88798e9ad8d251dfd76ff34be622c467b1b2d94a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
