)]}'
{"id":"openstack%2Foslo.vmware~725334","triplet_id":"openstack%2Foslo.vmware~master~I4ca5d293e1ca6fa3be52791c7ae2ffc612b329f3","project":"openstack/oslo.vmware","branch":"master","hashtags":[],"change_id":"I4ca5d293e1ca6fa3be52791c7ae2ffc612b329f3","subject":"Align contributing doc with oslo\u0027s policy","status":"MERGED","created":"2020-05-04 16:36:41.000000000","updated":"2020-06-30 16:10:37.000000000","submitted":"2020-06-30 16:07:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725334-1593533279264-2d1bbc0e","meta_rev_id":"df53b7ecf62517302f6425746e1638dc72e7166d","_number":725334,"virtual_id_number":725334,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 16:07:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-06-30 15:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-06-30 15:56:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 15:56:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-30 16:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a79371286dd830465da7f5cd1b84f4e7a7d005c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-04 16:36:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80f145ccc70cf8614f1201285b97ba1699cf7e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:44:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb508cb606e741b6980f0460994cbc32 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b7c401fcb0436a95b847803eff31bd : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"32391475a8cff8919ab44f95cc96ac7bfbd94506","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-05 12:33:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aac38a2e7b1f5e5a230071bd99db6f36665703cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:42:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0522abd074fb4847bed4e6207532c9f2 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46504d8aea7c464e843d1bcabe4a69f2 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"55ff232ba0d1a0d53466bd15428b01d5fe315b8d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-30 15:56:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTrivial doc change. Fast approving","accounts_in_message":[],"_revision_number":2},{"id":"4c8e3aa669d4cfa85f41e12e25c04dc91a336168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 15:56:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ec634177edee6ffe5b908e9c3482752a2443249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:07:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f12987e25174ebb905d11fb716d29f0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237d563410764f98b431af424ad70cdc : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a23d2315393afe115dce5315c1b9d465eb71c0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:07:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"df53b7ecf62517302f6425746e1638dc72e7166d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:10:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a311af58aea4c0d8ba1f67500e7c9d7 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab9ccf63a05da5f9bc326e1310f79d9fdbf97893","revisions":{"93548d6ee5cd5cfcc0472a2f19c0a80257dc032b":{"kind":"REWORK","_number":1,"created":"2020-05-04 16:36:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/34/725334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/34/725334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/1"}}},"commit":{"parents":[{"commit":"af6d26cee1a204d05c40ef9f8c5ac269d52a7e15","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/af6d26cee1a204d05c40ef9f8c5ac269d52a7e15"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:36:23.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:36:23.000000000","tz":120},"subject":"Align contributing doc with oslo\u0027s policy","message":"Align contributing doc with oslo\u0027s policy\n\nRelated to:\n- https://review.opendev.org/#/c/723044/\n- https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nChange-Id: I4ca5d293e1ca6fa3be52791c7ae2ffc612b329f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/93548d6ee5cd5cfcc0472a2f19c0a80257dc032b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/93548d6ee5cd5cfcc0472a2f19c0a80257dc032b"}]},"branch":"refs/heads/master"},"ab9ccf63a05da5f9bc326e1310f79d9fdbf97893":{"kind":"REWORK","_number":2,"created":"2020-05-05 12:33:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/34/725334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/34/725334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/34/725334/2"}}},"commit":{"parents":[{"commit":"af6d26cee1a204d05c40ef9f8c5ac269d52a7e15","subject":"Merge \"Add py38 package metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/af6d26cee1a204d05c40ef9f8c5ac269d52a7e15"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-04 16:36:23.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-05 12:32:54.000000000","tz":120},"subject":"Align contributing doc with oslo\u0027s policy","message":"Align contributing doc with oslo\u0027s policy\n\nRelated to:\n- https://review.opendev.org/#/c/723044/\n- https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\n\nChange-Id: I4ca5d293e1ca6fa3be52791c7ae2ffc612b329f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/ab9ccf63a05da5f9bc326e1310f79d9fdbf97893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/ab9ccf63a05da5f9bc326e1310f79d9fdbf97893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
