)]}'
{"id":"openstack%2Foslo.vmware~850234","triplet_id":"openstack%2Foslo.vmware~master~I188c80c3cb3c7a3d4aa54aea98ae37cdceabd427","project":"openstack/oslo.vmware","branch":"master","attention_set":{},"removed_from_attention_set":{"34602":{"account":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"last_update":"2022-07-19 12:23:15.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-07-19 08:40:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I188c80c3cb3c7a3d4aa54aea98ae37cdceabd427","subject":"Pin virtualenv","status":"ABANDONED","created":"2022-07-18 15:06:27.000000000","updated":"2022-07-19 13:05:30.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2616f5d517fcffbfda2eb29bab3aea1a75efe99e","_number":850234,"virtual_id_number":850234,"owner":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-18 16:45:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2022-07-19 08:40:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 15:08:50.000000000","updated_by":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-18 16:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91c006d3cc20cc0ad16f68f623e16c866d7d6e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"date":"2022-07-18 15:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c0f85a15e4a9c2a0715298e6a56ea20117c2760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f172e11e4064bf0854cdd52da6c8fe7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29f6f497a62849a4946ed995ce347990 : SUCCESS in 1h 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154753bc71b7445a86ebc82225734877 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a878005c59bb4629ac30aa41e27bf6c5 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdda34c7f3e24db3bf2080f44cb63f65 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c227e6d7e0c4b11a0957fd55822574c : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588647b36b104d5fba151a68538beda1 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6289dc8e8f6b4a3e9ca7963d77268b77 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5a6b22a1c5631aa48b216e26a69280167f801fbe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-19 08:40:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a41c86f2b2663959b184009bda2013983fd7d0d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"date":"2022-07-19 12:23:15.000000000","message":"Abandoned\n\nAbandoned in favor of uploading it to xena branch, where this issue remains unfixed.","accounts_in_message":[],"_revision_number":1},{"id":"2702dd4c5eb3b8c442de700734b3f783a0e7992b","author":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"date":"2022-07-19 12:56:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2616f5d517fcffbfda2eb29bab3aea1a75efe99e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-19 13:05:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fbd77c63703fbb10c6d542e0e2b2a590294c76d","revisions":{"3fbd77c63703fbb10c6d542e0e2b2a590294c76d":{"kind":"REWORK","_number":1,"created":"2022-07-18 15:06:27.000000000","uploader":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"ref":"refs/changes/34/850234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/34/850234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/850234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/850234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/34/850234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/34/850234/1"}}},"commit":{"parents":[{"commit":"88c2b2ede6ac514de1ee1b0671c587751916295c","subject":"Merge \"Fix formatting of release list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/88c2b2ede6ac514de1ee1b0671c587751916295c"}]}],"author":{"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","date":"2022-07-18 15:04:59.000000000","tz":120},"committer":{"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","date":"2022-07-18 15:06:26.000000000","tz":120},"subject":"Pin virtualenv","message":"Pin virtualenv\n\nThis change fixes the issue with suds-jurko incompatibility\nwith setuptools \u003e\u003d 58.0.0 by pinning virtualenv version.\n\nChange-Id: I188c80c3cb3c7a3d4aa54aea98ae37cdceabd427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/3fbd77c63703fbb10c6d542e0e2b2a590294c76d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/3fbd77c63703fbb10c6d542e0e2b2a590294c76d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
