)]}'
{"id":"openstack%2Foslo.vmware~850479","triplet_id":"openstack%2Foslo.vmware~stable%2Fwallaby~I472bcfd7b396c7567117a0d2a626235d471750e3","project":"openstack/oslo.vmware","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-07-20 13:52:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I472bcfd7b396c7567117a0d2a626235d471750e3","subject":"[stable-only] Pin virtualenv","status":"MERGED","created":"2022-07-20 10:29:52.000000000","updated":"2022-07-20 13:53:40.000000000","submitted":"2022-07-20 13:52:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850479","cherry_pick_of_change":850347,"cherry_pick_of_patch_set":1,"meta_rev_id":"1adcaca6fd39bd8ca602a814c004c80520d29b70","_number":850479,"virtual_id_number":850479,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 13:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-20 10:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-20 10:34:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 10:29:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":34602,"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","username":"ibernal"},"state":"CC"},{"updated":"2022-07-20 10:34:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-07-20 12:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97f2cafdf7d21d9587d39549a620cc6ea56fcea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-20 10:29:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"50ae76b89896ec55e183d923c24f4882140c12a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-20 10:34:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa6d857a28237a63ee5b2fbb93758ae0b648f92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:23:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cdcef637b544ad6b5117a30a1d919c6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/276d9f2666e5418f9dcab06efa868873 : SUCCESS in 1h 48m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f95e35863704084868a686b20d1f41e : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c631d549e764e4ab561d0bd888a9cec : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92c526d1555e4c26ae5bca0708f21d56 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a03ea41d92d84ab88a33697f6e255b74 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e70ce183974d09a7aae671e640e13e : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd1dc9f9f1054dbc9b9177017689f3ab : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ef2d02af41ad4da07bebc4db21820023b50244d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:24:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2850bb8c8a1d9d181ae51fd47e05376f3e8c97e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 13:52:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/556f0a38190f47babac3b6493361f9bf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20b0241176174671a96c2ec9331b1126 : SUCCESS in 1h 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d2e088827a49b38ef308c0b0a9da09 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ffaf91243e747df9e47f630d3aa88bc : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1073dc94cb47e08c63298391cd074f : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acebaec677204f018a6f883d3926adb2 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe55e9f41c08410fafe0e46ed269d75b : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ed10dba7fd5a4e626fedadb1af428a7dd2fe6831","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 13:52:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1adcaca6fd39bd8ca602a814c004c80520d29b70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 13:53:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3c3dc9bd5b64604ba77698d7e4c6915\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf28fd6002a8463c993c122f4ed1d99b : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed1edf94f9374887a5d5433cf66d02dc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc1a4668581ecb2724ecac007ff5baf6d47bc432","revisions":{"dc1a4668581ecb2724ecac007ff5baf6d47bc432":{"kind":"REWORK","_number":1,"created":"2022-07-20 10:29:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/79/850479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/79/850479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/850479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/850479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/79/850479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/79/850479/1"}}},"commit":{"parents":[{"commit":"27eeee2a81452d1fd5578e5abcbb2262092df42a","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/27eeee2a81452d1fd5578e5abcbb2262092df42a"}]}],"author":{"name":"Ignacio Horcada Bernal","email":"ihorcada@gmail.com","date":"2022-07-19 12:44:36.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-07-20 10:29:52.000000000","tz":0},"subject":"[stable-only] Pin virtualenv","message":"[stable-only] Pin virtualenv\n\nThis patch introduces pinning virtualenv version\nto ensure compatibility of setuptools with suds-jurko.\n\nChange-Id: I472bcfd7b396c7567117a0d2a626235d471750e3\n(cherry picked from commit 3a739bdd8510bbd18c0dc9a63735c875450cb0ea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/dc1a4668581ecb2724ecac007ff5baf6d47bc432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/dc1a4668581ecb2724ecac007ff5baf6d47bc432"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
