)]}'
{"id":"openstack%2Foslo.vmware~882193","triplet_id":"openstack%2Foslo.vmware~master~I77db3800c22bb8f04a33e6c79f2da8947f30d484","project":"openstack/oslo.vmware","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-08 09:18:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-08 10:34:49.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-05-08 10:34:49.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-08 10:34:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77db3800c22bb8f04a33e6c79f2da8947f30d484","subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","status":"MERGED","created":"2023-05-03 20:01:45.000000000","updated":"2023-05-08 10:35:52.000000000","submitted":"2023-05-08 10:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":879316,"submission_id":"882193-oslo-drop-py38","meta_rev_id":"d70b32944a86c1381094d93af488625c81261866","_number":882193,"virtual_id_number":882193,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-08 10:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-05-08 09:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-05-08 09:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-03 20:01:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-05-03 20:01:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-03 20:01:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-03 20:01:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a85f5de1b1bccfe0037918b3efc1171b6f566638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:01:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5846d7425ef16949287ddaf8e054d72c613ceba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 21:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27f91d45492842eaa5dd77fee6523728\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/188aa29c512e47b881840f4618bc4bf7 : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29b3d03bab31472db659337a5ccc52bc : FAILURE in 1h 08m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61753629f190426d9e82250878f6505a : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc386e9dab049a2985a9493665825a6 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/157a87b57c5e4f71927713eb6982896d : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb59982679c4551a8045378b05a955b : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"74f95de9a9e90a0874ef0aa869f49c7c580e4955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-04 20:20:44.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"acfc8d9bb48780994075b71f208ffc87de126ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:49:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842ee46fbbdc4c18af97b7b753b332e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b24e8770c85a417690d49d77eb04d292 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/567516b782ae4478be238f64b87a19b9 : SUCCESS in 1h 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5384001664634dafa753809c0cad2117 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fccc3b30e57485688a083591420c3c9 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/726eb3d43d084017958392900828a85b : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f72e466ea341488f87975caff58af5 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1fcdbbe2fed7bd23d4e05b37c74229af636f27c2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-05 12:09:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e74a6107ee2bee19f0ff00cb3a01fafce9d6103a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-05 18:53:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2cc1fcf6fcbf7aabe309c75ec8939fa6f37139b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 20:29:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06e761dce6114348be8d16365ba16b6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5651dad082f4ecbb9a8ca35f614299d : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad98ca7a377e4500b12090fed902dc06 : SUCCESS in 1h 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071f9ef5c27d489383c39ff5edc5486e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7730bffb7484e76b1557582dab50ba1 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5962ea4b891e4da08a8a9eea5b526452 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc0b9de0ccb456f84cc736cd5ad0e56 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"475e19acbae49ad98b4a29e0295560b500b9ec41","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-08 09:18:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c14c112de81e7d1b24204bfeffb39b1f1275973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 09:19:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"632eeb6424e396e17f0f4de8a6a05ce5df02f483","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 10:34:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8822be63420f4517b27ee44389880759\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a20393ee36544c1bbf9c2c10af56a2f : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf5df000a56b415dbfe7075b8956b101 : SUCCESS in 1h 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11b7ac4083a4d928d639fa4e4c94a4f : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74b65617ca2e4cda9546b3225ba6fcec : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56cc244fbaf94408bbaf6bb44430bc6d : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d7b0eb44c443cb8335aa2832c3f59c : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"16304d829fa32760469a563e4a2c02939bb359fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 10:34:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d70b32944a86c1381094d93af488625c81261866","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 10:35:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a22aa2de5e948f48b1a4e95acbb2df7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b526864319314b65966967ccd3ccc814 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f442d815c705c126b1bcdd89024ebd3160328491","revisions":{"23fb323e70671557b3e12d2c60b09abfa0ffa8d2":{"kind":"REWORK","_number":1,"created":"2023-05-03 20:01:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/882193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/93/882193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/1"}}},"commit":{"parents":[{"commit":"d6cdf97c4559cfde324289d191460470ee96ef02","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d6cdf97c4559cfde324289d191460470ee96ef02"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:45.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:45.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit d6cdf97c4559cfde324289d191460470ee96ef02.\n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: I77db3800c22bb8f04a33e6c79f2da8947f30d484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/23fb323e70671557b3e12d2c60b09abfa0ffa8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/23fb323e70671557b3e12d2c60b09abfa0ffa8d2"}]},"branch":"refs/heads/master"},"eb10acb026625b527f838ab38da536c683aab2b3":{"kind":"REWORK","_number":2,"created":"2023-05-04 20:20:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/882193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/93/882193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/2"}}},"commit":{"parents":[{"commit":"d6cdf97c4559cfde324289d191460470ee96ef02","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d6cdf97c4559cfde324289d191460470ee96ef02"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:45.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:20:44.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit d6cdf97c4559cfde324289d191460470ee96ef02.\n\nKeeping Python 3.10 in setup.cfg classifier. \n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: I77db3800c22bb8f04a33e6c79f2da8947f30d484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/eb10acb026625b527f838ab38da536c683aab2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/eb10acb026625b527f838ab38da536c683aab2b3"}]},"branch":"refs/heads/master"},"f442d815c705c126b1bcdd89024ebd3160328491":{"kind":"REWORK","_number":3,"created":"2023-05-05 18:53:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/882193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/93/882193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/93/882193/3"}}},"commit":{"parents":[{"commit":"d6cdf97c4559cfde324289d191460470ee96ef02","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/d6cdf97c4559cfde324289d191460470ee96ef02"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-03 20:01:45.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-05 18:53:04.000000000","tz":0},"subject":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"","message":"Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n\nThis reverts commit d6cdf97c4559cfde324289d191460470ee96ef02.\n\nKeeping Python 3.10 in setup.cfg classifier and zuul.yaml changes.\n\nReason for revert:\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nTC has been discussing about re-adding the python 3.8\ntesting in current master 2023.2 release testing.\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-04-25-18.00.log.html#l-191\n- https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033469.html\n\nWhile governance changes are under review, TC agreed to add py3.8 testing\nso that we do not see more project/lib dropping python 3.8 and make them\nuninstalable on python 3.8\n\n- https://meetings.opendev.org/meetings/tc/2023/tc.2023-05-02-18.00.log.html#l-17\n- https://review.opendev.org/c/openstack/governance/+/882165\n\nAlso adding py3.8 testing back in job https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/882175\n\nChange-Id: I77db3800c22bb8f04a33e6c79f2da8947f30d484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/f442d815c705c126b1bcdd89024ebd3160328491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/f442d815c705c126b1bcdd89024ebd3160328491"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
