)]}'
{"id":"openstack%2Foslo.vmware~960655","triplet_id":"openstack%2Foslo.vmware~master~I76d9e3171d6602cd7556014678e0356730fb2743","project":"openstack/oslo.vmware","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-12 17:48:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76d9e3171d6602cd7556014678e0356730fb2743","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-09-11 17:19:20.000000000","updated":"2025-09-12 17:49:30.000000000","submitted":"2025-09-12 17:48:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960655-pyproject","meta_rev_id":"db504091deb42d82366d1582c323f9a26cfb1d23","_number":960655,"virtual_id_number":960655,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 17:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-09-12 14:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-09-12 14:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 19:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 14:12:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"6d2af0afdf3acd442863c2242062ebd3fae6cbed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-11 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c52abb74e058107d6a36052606a738009cfeba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:09:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6f1a4c2b4c4b63865ef36a9c3a7735\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11e90dc60a4145afa86e36bf5312ced7 : SUCCESS in 2m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc44af80fc17465190a203c9da0d5227 : SUCCESS in 1h 44m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228c49edd0d84dc9be67e753d87680b4 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ccddecc5cbe4e12a7ef92ab98f98376 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80d03699b7564a29bcd33a8b98e2fb1d : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f37d872d4990458b84bbd9ab2258268f : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf91922a158843308dd634ae2eb4955e : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"43e603bbb5554ac5db40958b9b1c175315ee5729","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 14:12:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"657e9ffa17b4d5c2b415953e54a2b69685bb3281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 14:12:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46839a8e7b42b66670716fb806722314d019e051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 14:58:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3423695520064fb69ecf55bcbf09cd81\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/71a0c0222a93484e9b469226ce7ca4b5 : SUCCESS in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7bf3eb50bd44900aba97521cd37c7a2 : FAILURE in 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a8429ba33a46c3a60b82f86e0fdc30 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80bb1502f5a9421ea9b9a0ab80e71e39 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/824f51cb041d45af9c498e9015feb915 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4c054031ef4f33967e6231dddafb85 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2cf6187830b5e2afd2f1d235aa1b361b292c167f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-12 15:03:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa852f3e83d4dd9f2d775ed97ac029c081e5c8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 16:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7ff932cecf4bb597a9254443cdb2d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a29ee56950224349ae69a804a3806f12 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a667c462ec948acb45cdbc331ff215b : SUCCESS in 1h 40m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c354c3942344ef92b00f306594e087 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b05ab429c194532955be157b03b8eff : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45c63caf0cde492893ae3106795744b9 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9d87d3f7eb44099ac207b05c6d3816a : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2360050b2a1040c5998fcba07e847ab5 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"761ee646f560f619b4db9baa6a6393df2d8279b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 16:49:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c13973e3a8ec0a5cb8dace920666847b7bf774a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:48:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70fea4221f24908b51ab9f5e6ac335e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca0eb43823244c21a9012e85bce41c98 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ae2283632094f36ab32a82198c38036 : SUCCESS in 58m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8472e3415f54e6fb0a32573b77b4da2 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16fc66a839d14d37832cd6483cef73a5 : SUCCESS in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c99b395e4d44eba83b1fa982911d290 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebefc8a35674931aeefbf613bf656cc : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ff715c149353138baf670a5d41406b32e51eecc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:48:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db504091deb42d82366d1582c323f9a26cfb1d23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bcad44a44604b9385aec3c00ac92195\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a4a9f488064c1b90c4afb36d145005 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d79c704bbe6873bffbb85d3562d0c6fa5051609","revisions":{"1d79c704bbe6873bffbb85d3562d0c6fa5051609":{"kind":"REWORK","_number":1,"created":"2025-09-11 17:19:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/960655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.vmware","ref":"refs/changes/55/960655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/55/960655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/55/960655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.vmware refs/changes/55/960655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.vmware refs/changes/55/960655/1"}}},"commit":{"parents":[{"commit":"201ea4bfa53185787ffe6b2921a1040c351ba692","subject":"Drop Python 3.9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/201ea4bfa53185787ffe6b2921a1040c351ba692"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 17:18:11.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 17:18:48.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: I76d9e3171d6602cd7556014678e0356730fb2743\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/1d79c704bbe6873bffbb85d3562d0c6fa5051609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.vmware/commit/1d79c704bbe6873bffbb85d3562d0c6fa5051609"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
