)]}'
{"id":"openstack%2Fosops~823687","triplet_id":"openstack%2Fosops~master~I97f46310cbf4935f113100f0e10a4e2965eee714","project":"openstack/osops","branch":"master","topic":"fixglancelist","attention_set":{"16339":{"account":{"_account_id":16339,"name":"Cyril Lopez","email":"cylopez@redhat.com","username":"cylopez"},"last_update":"2024-06-13 19:50:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-06-13 19:50:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}},"hashtags":[],"change_id":"I97f46310cbf4935f113100f0e10a4e2965eee714","subject":"[fix] glance nagios","status":"NEW","created":"2022-01-06 13:08:17.000000000","updated":"2024-06-13 19:50:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1f65fe2e4b443263e5417368a264572b538bc7d","_number":823687,"virtual_id_number":823687,"owner":{"_account_id":16339,"name":"Cyril Lopez","email":"cylopez@redhat.com","username":"cylopez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-06 13:09:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-06-13 19:50:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-06 19:10:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2024-06-13 19:50:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 13:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 19:10:01.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2022-01-07 09:43:47.000000000","updated_by":{"_account_id":16339,"name":"Cyril Lopez","email":"cylopez@redhat.com","username":"cylopez"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"d88b19c2177a45f975029d707fcbe969dc1aac2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16339,"name":"Cyril Lopez","email":"cylopez@redhat.com","username":"cylopez"},"date":"2022-01-06 13:08:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40a568c0822dce0ea268dfe835f26645089e867b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:09:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/0fbca69fae7d4dcd9036d9e03c5e5a6d : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b10eb00bbb8d20e5f787680dbd317250844cfa45","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-01-06 19:10:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1f65fe2e4b443263e5417368a264572b538bc7d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-06-13 19:50:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e37254d50ccafc7c9cac9b5ebffea87e3ba49c56","revisions":{"e37254d50ccafc7c9cac9b5ebffea87e3ba49c56":{"kind":"REWORK","_number":1,"created":"2022-01-06 13:08:17.000000000","uploader":{"_account_id":16339,"name":"Cyril Lopez","email":"cylopez@redhat.com","username":"cylopez"},"ref":"refs/changes/87/823687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osops","ref":"refs/changes/87/823687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osops refs/changes/87/823687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osops refs/changes/87/823687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osops refs/changes/87/823687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osops refs/changes/87/823687/1"}}},"commit":{"parents":[{"commit":"1ac9d08f9c6ffbf94bab4a1c4bf06ef65348ecdd","subject":"[fix] list-vms-by-host obsolet command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osops/commit/1ac9d08f9c6ffbf94bab4a1c4bf06ef65348ecdd"}]}],"author":{"name":"Cyril Lopez","email":"cylopez@redhat.com","date":"2022-01-06 13:04:52.000000000","tz":60},"committer":{"name":"Cyril Lopez","email":"cylopez@redhat.com","date":"2022-01-06 13:07:29.000000000","tz":60},"subject":"[fix] glance nagios","message":"[fix] glance nagios\n\nglance cli is now replaced by openstack image.\n\nTested on Xena\n\nStory: 2008299\nTask: 41184\n\nChange-Id: I97f46310cbf4935f113100f0e10a4e2965eee714\nSigned-off-by: Cyril Lopez \u003ccylopez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osops/commit/e37254d50ccafc7c9cac9b5ebffea87e3ba49c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osops/commit/e37254d50ccafc7c9cac9b5ebffea87e3ba49c56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
