)]}'
{"id":"openstack%2Fosprofiler~1005499","triplet_id":"openstack%2Fosprofiler~master~Ifeab1196ece12218cfcaa9dcd46a6b02ba6524e4","project":"openstack/osprofiler","branch":"master","hashtags":[],"change_id":"Ifeab1196ece12218cfcaa9dcd46a6b02ba6524e4","subject":"Ensure recursive detection of package files","status":"MERGED","created":"2026-09-14 05:10:31.000000000","updated":"2026-09-14 08:53:26.000000000","submitted":"2026-09-14 08:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005499","meta_rev_id":"3b21297e9f2115f4a8cf2ee060c8e6242ef2da80","_number":1005499,"virtual_id_number":1005499,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 08:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-14 07:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-14 07:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 06:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f5e3a2615257f1bd1f45a83702b2844e8bc7b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 05:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e822a339765f9b735c64e20cc6ad5eb4945b236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 06:16:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eda2a49cc404f3cb9ed8c463a215965\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a86e0a0fdf0497398b845063a81d255 : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/626af80bd79745deb26e7eabff929e38 : SUCCESS in 3m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc03b3015994494ca569e13b591523bd : SUCCESS in 1h 04m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bae0a5155d4469a8c8876cc47ea0198 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571dc8d76536480483e7addd8b59d878 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d81d4c68d6b9445596b1c31190d1de4e : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf6471d73cd5401b8ee9c4756e8ce7dd : SUCCESS in 7m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a18196f9f22422381c814e5f993b7fe : SUCCESS in 5m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7486648a299f469eb72e4211e2e095d1 : SUCCESS in 14m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4691ffc8119c4f1681fd80ad34d4834f : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/441a76d894fe4cd3894ac0b3e072f440 : SUCCESS in 3m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/86ec76d0905d411c941ed42f941a6b70 : SUCCESS in 2m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/94abe64870a8439cbaf210809ffdb1c4 : SUCCESS in 5m 36s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/3190a1036ac8414da4d2f0cbab34304b : POST_FAILURE in 25m 03s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/165dca0c63a74f18bccd09c7a96138ca : POST_FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85976edfae862c7284b90bfcd917ec3294a7b553","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 07:39:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4848fea4b97c4d31bfe3b4eb3f89a2127a1bfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 07:40:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2fa5b43a3ebfe613ef55eef31006502dda0d244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:52:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534d2d1159864b9996f46b802760c22c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43599ea8c721466497d495420c47c015 : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5b92451d3324ba7ade73a3d088c81ad : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e810a7e6fdd4007bf9cd64882e9e94c : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876c841a19c74e4083c3a08fbb289be8 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/215cf95ad2124c2b8bbf3bbc493bc869 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce03ac0d983e4ca1baa4b4f3cd643aa9 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c7035a0b9d14407aa060187904eba3b : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078c4de1eed34ace82385febacf3b01d : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6c5ff99da0f499a9f89d1b295a33dff : SUCCESS in 3m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/054b6f1d1e18488db2aaa1543963cffe : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8769fb00f6bddef6867b3415a4c8f8f437fa7487","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:52:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b21297e9f2115f4a8cf2ee060c8e6242ef2da80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88caa5abbce49ab91849dd9b798d3b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a7a725574841c384c408b52f3ae4ef : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9798ff7919fc033ee0858cb9de410334743ab657","revisions":{"9798ff7919fc033ee0858cb9de410334743ab657":{"kind":"REWORK","_number":1,"created":"2026-09-14 05:10:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/1005499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/99/1005499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/99/1005499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/99/1005499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/99/1005499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/99/1005499/1"}}},"commit":{"parents":[{"commit":"7c80e3cb6639cec11fa7f768805939ed27c47078","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7c80e3cb6639cec11fa7f768805939ed27c47078"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 05:10:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 05:10:19.000000000","tz":540},"subject":"Ensure recursive detection of package files","message":"Ensure recursive detection of package files\n\nChange-Id: Ifeab1196ece12218cfcaa9dcd46a6b02ba6524e4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9798ff7919fc033ee0858cb9de410334743ab657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9798ff7919fc033ee0858cb9de410334743ab657"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
