)]}'
{"id":"openstack%2Fosprofiler~625198","triplet_id":"openstack%2Fosprofiler~master~I09a122f91e3d26531965fe53d6df7c66f502de4c","project":"openstack/osprofiler","branch":"master","topic":"improved-redis-driver","hashtags":[],"change_id":"I09a122f91e3d26531965fe53d6df7c66f502de4c","subject":"Optimize storage schema for Redis driver","status":"MERGED","created":"2018-12-14 09:51:54.000000000","updated":"2019-05-20 09:36:26.000000000","submitted":"2019-05-20 09:36:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"625198-1558344986401-38c90c1d","meta_rev_id":"c9c85717b73351e70e8e97bb55a849fd8017ece9","_number":625198,"virtual_id_number":625198,"owner":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-09 11:30:37.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-05-20 09:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 06:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":2,"date":"2019-05-20 07:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":1,"date":"2019-05-20 07:43:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-17 16:24:26.000000000","updated_by":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"reviewer":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"state":"REVIEWER"},{"updated":"2019-04-09 11:30:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 06:44:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-20 09:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b08da7810ba07ce4ac1672899f1e1c37319d4a5b","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2018-12-14 09:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98fce076bd3d996837b0ae8bc2834c80ed635bdc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-12-14 09:53:11.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/1/296/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"5910e05cd71a46b03cbd0d39d94347c93a01eecc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 11:15:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/98/625198/1/check/tempest-full/a2c13fe/ : SUCCESS in 1h 21m 47s\n- tempest-full-py3 http://logs.openstack.org/98/625198/1/check/tempest-full-py3/df0eac6/ : SUCCESS in 1h 10m 25s\n- openstack-tox-cover http://logs.openstack.org/98/625198/1/check/openstack-tox-cover/1818fc9/ : FAILURE in 3m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/1/check/openstack-tox-lower-constraints/f564408/ : FAILURE in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/1/check/openstack-tox-pep8/be71f83/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/1/check/openstack-tox-py27/ec09f03/ : FAILURE in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/98/625198/1/check/openstack-tox-py35/b0f4954/ : FAILURE in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/1/check/openstack-tox-py36/70284a6/ : FAILURE in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/98/625198/1/check/openstack-tox-docs/2c063d5/html/ : SUCCESS in 2m 43s\n- openstack-tox-functional http://logs.openstack.org/98/625198/1/check/openstack-tox-functional/02729dc/ : FAILURE in 3m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/1/check/openstack-tox-functional-py35/0732c7b/ : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e477f9c22a1d3609bce3014f86a5f435fb640424","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2018-12-14 19:03:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b659aec1a36864c3d3e9e8c1d32cb9812601de09","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2018-12-17 09:36:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"148a6891f53b951392fd592d8db85d3e6728236a","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2018-12-17 16:24:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"247963a8f88be0d9ba98560475afe695ace2a5fe","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-01-08 14:24:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd65c73b8c8305f310837e5d32115755f9c1684d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-08 14:25:44.000000000","message":"Patch Set 2:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/2/300/ : SUCCESS\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"e92a8e3ca2b07575f8e71d5ef15aa42c091f8ce8","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-01-08 14:49:49.000000000","message":"Patch Set 2: Workflow-1\n\ntesting...","accounts_in_message":[],"_revision_number":2},{"id":"0c2d00be6df64137ca3693002a97482143457b66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 15:49:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/2/check/tempest-full/3007c6d/ : SUCCESS in 1h 24m 07s\n- tempest-full-py3 http://logs.openstack.org/98/625198/2/check/tempest-full-py3/ae89c53/ : SUCCESS in 1h 17m 54s\n- openstack-tox-cover http://logs.openstack.org/98/625198/2/check/openstack-tox-cover/cbe285b/cover/ : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/2/check/openstack-tox-lower-constraints/ed477d8/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/2/check/openstack-tox-pep8/9a89f16/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/2/check/openstack-tox-py27/b3769b1/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/98/625198/2/check/openstack-tox-py35/227bb3b/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/2/check/openstack-tox-py36/b0f1c9b/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/98/625198/2/check/openstack-tox-docs/f4b6e1c/html/ : SUCCESS in 3m 02s\n- openstack-tox-functional http://logs.openstack.org/98/625198/2/check/openstack-tox-functional/d47ba54/ : SUCCESS in 4m 18s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/2/check/openstack-tox-functional-py35/f1c87e2/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"51b46cb90722febec20a78c466beda4d73578914","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-01-09 08:25:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b3e43185f8400386cfa64329a4dee99e49650417","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-09 08:28:31.000000000","message":"Patch Set 3:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/3/303/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"8aefc2778a88ebcb08f28c879d2dc48c64993dd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 09:50:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/3/check/tempest-full/a23d4e7/ : SUCCESS in 1h 15m 42s\n- tempest-full-py3 http://logs.openstack.org/98/625198/3/check/tempest-full-py3/619b483/ : SUCCESS in 1h 19m 48s\n- openstack-tox-cover http://logs.openstack.org/98/625198/3/check/openstack-tox-cover/1f9fb4a/cover/ : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/3/check/openstack-tox-lower-constraints/3ddb6a4/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/3/check/openstack-tox-pep8/9b1b22b/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/3/check/openstack-tox-py27/5ec2493/ : SUCCESS in 3m 06s\n- openstack-tox-py35 http://logs.openstack.org/98/625198/3/check/openstack-tox-py35/fd59d7c/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/3/check/openstack-tox-py36/a52ffdd/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/98/625198/3/check/openstack-tox-docs/fa615ea/html/ : SUCCESS in 2m 59s\n- openstack-tox-functional http://logs.openstack.org/98/625198/3/check/openstack-tox-functional/3ce59f3/ : SUCCESS in 3m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/3/check/openstack-tox-functional-py35/d9cdb27/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"27ace9b8b51ec2333e979c097ac702d6d0d04bdc","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-01-09 13:02:06.000000000","message":"Patch Set 3:\n\nverified with Rally in https://review.openstack.org/#/c/615350/34 \nreport: http://logs.openstack.org/50/615350/34/check/neutron-rally-task/c573df5/results/workloads/0_NeutronNetworks_create_and_list_networks.html#/NeutronNetworks.create_and_list_networks/output","accounts_in_message":[],"_revision_number":3},{"id":"d6fcbffc0bf137a49a30872aef6878ce5b982d65","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-03-18 16:26:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f7bf7ba7b0c597aff58605b02b3f786b8032cb82","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 16:29:16.000000000","message":"Patch Set 4:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/4/320/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":4},{"id":"26592dd57a47d7d9942bbe90e6a4f65ba45fb60a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 18:07:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/4/check/tempest-full/7d2a298/ : SUCCESS in 1h 34m 19s\n- tempest-full-py3 http://logs.openstack.org/98/625198/4/check/tempest-full-py3/d9ecd18/ : SUCCESS in 1h 33m 21s\n- openstack-tox-cover http://logs.openstack.org/98/625198/4/check/openstack-tox-cover/3e8f7ab/cover/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/4/check/openstack-tox-lower-constraints/bffe404/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/4/check/openstack-tox-pep8/6acf159/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/4/check/openstack-tox-py27/8e23d1c/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/4/check/openstack-tox-py36/a267b38/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/4/check/openstack-tox-py37/a13b3ae/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/98/625198/4/check/openstack-tox-docs/30b9965/html/ : SUCCESS in 4m 48s\n- openstack-tox-functional http://logs.openstack.org/98/625198/4/check/openstack-tox-functional/f59a288/ : SUCCESS in 5m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/4/check/openstack-tox-functional-py35/0a5af1e/ : SUCCESS in 6m 05s\n- tempest-full-py3-profiler-enabled http://logs.openstack.org/98/625198/4/check/tempest-full-py3-profiler-enabled/c886f05/ : SUCCESS in 48m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37073a9d5915820fe6d45dd3d00cb35b0c770007","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-03-22 16:17:38.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"57d73dd91c15819ff806ec611b17e49cf69d91f4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-22 16:20:44.000000000","message":"Patch Set 5:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/5/323/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":5},{"id":"cdd188579dc61eb42813b819e41d47ebf63a3e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 18:13:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/5/check/tempest-full/38666da/ : SUCCESS in 1h 43m 24s\n- tempest-full-py3 http://logs.openstack.org/98/625198/5/check/tempest-full-py3/4298a19/ : SUCCESS in 1h 29m 29s\n- openstack-tox-cover http://logs.openstack.org/98/625198/5/check/openstack-tox-cover/1e313bb/cover/ : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/5/check/openstack-tox-lower-constraints/50e30b5/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/5/check/openstack-tox-pep8/2a572c6/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/5/check/openstack-tox-py27/dbf3462/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/5/check/openstack-tox-py36/2b000c3/ : SUCCESS in 4m 57s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/5/check/openstack-tox-py37/ea2d01f/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/98/625198/5/check/openstack-tox-docs/8513526/html/ : SUCCESS in 4m 03s\n- openstack-tox-functional http://logs.openstack.org/98/625198/5/check/openstack-tox-functional/17120de/ : SUCCESS in 4m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/5/check/openstack-tox-functional-py35/fa20996/ : SUCCESS in 4m 51s\n- tempest-full-py3-profiler-enabled http://logs.openstack.org/98/625198/5/check/tempest-full-py3-profiler-enabled/5d825b9/ : SUCCESS in 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db63f7be89ae46d95b847b74c64aa4267f0efdf9","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-04-08 10:21:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e194ab24371d3232c753f71a43f78035729ddf7b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-08 10:24:54.000000000","message":"Patch Set 6:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/6/328/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":6},{"id":"c1a8cc07421fabb038d5782c3dedd0450e825ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 12:01:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/98/625198/6/check/tempest-full/b78eed3/ : FAILURE in 1h 32m 17s\n- tempest-full-py3 http://logs.openstack.org/98/625198/6/check/tempest-full-py3/5032d93/ : SUCCESS in 1h 32m 03s\n- openstack-tox-cover http://logs.openstack.org/98/625198/6/check/openstack-tox-cover/72271d6/cover/ : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/6/check/openstack-tox-lower-constraints/3bb273f/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/6/check/openstack-tox-pep8/5d14a8d/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/6/check/openstack-tox-py27/20e962e/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/6/check/openstack-tox-py36/7c6430d/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/6/check/openstack-tox-py37/1c793a1/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/98/625198/6/check/openstack-tox-docs/f7ba584/html/ : SUCCESS in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/98/625198/6/check/build-openstack-releasenotes/500a578/html/ : SUCCESS in 2m 48s\n- openstack-tox-functional http://logs.openstack.org/98/625198/6/check/openstack-tox-functional/7ac2080/ : SUCCESS in 3m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/6/check/openstack-tox-functional-py35/67bdf97/ : SUCCESS in 3m 39s\n- tempest-full-py3-profiler-enabled http://logs.openstack.org/98/625198/6/check/tempest-full-py3-profiler-enabled/7f089bc/ : SUCCESS in 41m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4d1c4e77de2e0a7b9315b10733170a85e73e1b0f","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-04-08 12:49:40.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"033196d5267fec88eb68f430a57d4e96964de972","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-08 12:52:42.000000000","message":"Patch Set 6:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/6/329/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":6},{"id":"3f1c47e7369fa7daaa6229750797d1edf742fb7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 14:43:11.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/98/625198/6/check/tempest-full/8ce5f3a/ : FAILURE in 1h 26m 31s\n- tempest-full-py3 http://logs.openstack.org/98/625198/6/check/tempest-full-py3/0e75792/ : SUCCESS in 1h 46m 49s\n- openstack-tox-cover http://logs.openstack.org/98/625198/6/check/openstack-tox-cover/e75f03c/cover/ : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/6/check/openstack-tox-lower-constraints/3685be4/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/6/check/openstack-tox-pep8/6d20e0d/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/6/check/openstack-tox-py27/9215948/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/6/check/openstack-tox-py36/a52a146/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/6/check/openstack-tox-py37/61b656b/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/98/625198/6/check/openstack-tox-docs/5549d65/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/98/625198/6/check/build-openstack-releasenotes/9478c5b/html/ : SUCCESS in 2m 59s\n- openstack-tox-functional http://logs.openstack.org/98/625198/6/check/openstack-tox-functional/54c1dc8/ : SUCCESS in 3m 51s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/6/check/openstack-tox-functional-py35/a0ab602/ : SUCCESS in 4m 42s\n- tempest-full-py3-profiler-enabled http://logs.openstack.org/98/625198/6/check/tempest-full-py3-profiler-enabled/dba6378/ : SUCCESS in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"56664c4c96416788af8cac7721e993515470f005","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-04-09 11:27:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7b3d0034cbbcbe137a886827cb4d13c2f1a6a3b6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-09 11:30:37.000000000","message":"Patch Set 7:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/625198/7/337/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":7},{"id":"c0f751b1ac8014d91379a6a2bb2645d48ee83596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 13:26:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/7/check/tempest-full/c9d69d8/ : SUCCESS in 1h 53m 45s\n- tempest-full-py3 http://logs.openstack.org/98/625198/7/check/tempest-full-py3/700104f/ : SUCCESS in 1h 37m 34s\n- openstack-tox-cover http://logs.openstack.org/98/625198/7/check/openstack-tox-cover/aef2944/cover/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/7/check/openstack-tox-lower-constraints/431e722/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/7/check/openstack-tox-pep8/d877da3/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/7/check/openstack-tox-py27/05a0ce6/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/7/check/openstack-tox-py36/55cd3d1/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/7/check/openstack-tox-py37/e1804b0/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/98/625198/7/check/openstack-tox-docs/5a69e09/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes http://logs.openstack.org/98/625198/7/check/build-openstack-releasenotes/23e5533/html/ : SUCCESS in 3m 14s\n- openstack-tox-functional http://logs.openstack.org/98/625198/7/check/openstack-tox-functional/4e816dd/ : SUCCESS in 4m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/625198/7/check/openstack-tox-functional-py35/c7bbff1/ : SUCCESS in 5m 07s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/98/625198/7/check/tempest-smoke-py3-osprofiler-redis/23b148d/ : SUCCESS in 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"078960ed16af141dfa08037ae6779964619d364d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-20 06:44:49.000000000","message":"Patch Set 7: Code-Review+1\n\nHi Ilya, do You think this patch is ready to be merged? All pieces on rally side are already merged, Andrey did great work to make this integration working fine. So when this patch would be merged (and released) I think we will be able to move forward with https://review.opendev.org/#/c/615350/ and use osprofiler reports in every neutron-rally job :)","accounts_in_message":[],"_revision_number":7},{"id":"8853af09487a66a3a192ec1616cd4eb3670ee4fd","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-20 07:42:35.000000000","message":"Patch Set 7:\n\n\u003e Hi Ilya, do You think this patch is ready to be merged? All pieces\n \u003e on rally side are already merged, Andrey did great work to make\n \u003e this integration working fine. So when this patch would be merged\n \u003e (and released) I think we will be able to move forward with\n \u003e https://review.opendev.org/#/c/615350/ and use osprofiler reports\n \u003e in every neutron-rally job :)\n\nHi Slawek, yes, the patch could be merged. The code is tested by functional tests, tempest job and rally, - so should be pretty safe.\nCool to see osprofiler reports in Neutron gate :)","accounts_in_message":[],"_revision_number":7},{"id":"6c68db73526b769055bc6de94d731366536f935a","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-20 07:43:35.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cf4ee44b4f388f64028dfc118adb64ee5a392bf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:43:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5a57d696d1930d2937290b6cca40ea3e61c8dcb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:36:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/98/625198/7/gate/tempest-full/e7a6fc2/ : SUCCESS in 1h 51m 46s\n- tempest-full-py3 http://logs.openstack.org/98/625198/7/gate/tempest-full-py3/26eb427/ : SUCCESS in 1h 32m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/625198/7/gate/openstack-tox-lower-constraints/bde61bd/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/98/625198/7/gate/openstack-tox-pep8/af8c8b8/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/98/625198/7/gate/openstack-tox-py27/98f2b28/ : SUCCESS in 2m 31s\n- openstack-tox-py36 http://logs.openstack.org/98/625198/7/gate/openstack-tox-py36/30320f1/ : SUCCESS in 2m 53s\n- openstack-tox-py37 http://logs.openstack.org/98/625198/7/gate/openstack-tox-py37/351db4a/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/98/625198/7/gate/openstack-tox-docs/61203a7/html/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/98/625198/7/gate/build-openstack-releasenotes/b2c986e/html/ : SUCCESS in 2m 16s\n- openstack-tox-functional http://logs.openstack.org/98/625198/7/gate/openstack-tox-functional/0e52477/ : SUCCESS in 2m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/98/625198/7/gate/openstack-tox-functional-py36/6ecfcae/ : SUCCESS in 3m 09s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/98/625198/7/gate/tempest-smoke-py3-osprofiler-redis/77f6e78/ : SUCCESS in 39m 56s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/98/625198/7/gate/tempest-smoke-py3-osprofiler-sqlalchemy/be7bfe7/ : SUCCESS in 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e022fdae84c8aa4ba55fa9c09b32c40483e1587a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:36:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"147effea78e92ae341fd048b1c9c1ff54c0ced0f","revisions":{"c6d5a5499ad482acaff13a98fd3e65f98d3e02e9":{"kind":"REWORK","_number":1,"created":"2018-12-14 09:51:54.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/1"}}},"commit":{"parents":[{"commit":"1657840f9fd87074c0a20f7f323ca10dd988ef4e","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1657840f9fd87074c0a20f7f323ca10dd988ef4e"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"subject":"[DNM] Try more optimal schema of Redis storage","message":"[DNM] Try more optimal schema of Redis storage\n\nThe original Redis driver stores each tracing event under its own key,\nas result list and get operations require full scan of the database.\nThe idea behind this patch is to store traces as Redis lists under key equal\nto trace id. In this case list operation iterates only over unique trace ids\nand get operation retrieves content of specified list. Note that list operation\nstill needs to retrieve at least 1 event from the trace to get timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c6d5a5499ad482acaff13a98fd3e65f98d3e02e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c6d5a5499ad482acaff13a98fd3e65f98d3e02e9"}]},"branch":"refs/heads/master"},"9d5004f3f150de93d2e6d91d2ec8fe4c2165ef46":{"kind":"REWORK","_number":2,"created":"2019-01-08 14:24:24.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/2"}}},"commit":{"parents":[{"commit":"1657840f9fd87074c0a20f7f323ca10dd988ef4e","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1657840f9fd87074c0a20f7f323ca10dd988ef4e"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-01-08 14:24:13.000000000","tz":60},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stores each tracing event under its own key,\nas result both list and get operations require full scan of the database.\nThe idea behind this patch is to store traces as Redis lists under key equal\nto trace id. In this case list operation iterates only over unique trace ids\nand get operation retrieves content of specified list. Note that list operation\nstill needs to retrieve at least 1 event from the trace to get timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9d5004f3f150de93d2e6d91d2ec8fe4c2165ef46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9d5004f3f150de93d2e6d91d2ec8fe4c2165ef46"}]},"branch":"refs/heads/master"},"0717beef9818b5b00c3d02b0c2fc9f405a0d453d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-01-09 08:25:08.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/3"}}},"commit":{"parents":[{"commit":"0b6b5bbcefd3f59ffc64f4e8818308da4e716a65","subject":"Allow test path to be overridden","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0b6b5bbcefd3f59ffc64f4e8818308da4e716a65"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-01-09 08:24:47.000000000","tz":60},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stores each tracing event under its own key,\nas result both list and get operations require full scan of the database.\nThe idea behind this patch is to store traces as Redis lists under key equal\nto trace id. In this case list operation iterates only over unique trace ids\nand get operation retrieves content of specified list. Note that list operation\nstill needs to retrieve at least 1 event from the trace to get timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0717beef9818b5b00c3d02b0c2fc9f405a0d453d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0717beef9818b5b00c3d02b0c2fc9f405a0d453d"}]},"branch":"refs/heads/master"},"9d50e6b86e6ed30f26f8f8ce6bb68dd8839663d2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-18 16:26:11.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/4"}}},"commit":{"parents":[{"commit":"b844b96bffdeebd111e0bbd299c4b47882cc8f63","subject":"Merge \"Collect traces from Tempest job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/b844b96bffdeebd111e0bbd299c4b47882cc8f63"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-03-18 16:26:11.000000000","tz":0},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stores each tracing event under its own key,\nas result both list and get operations require full scan of the database.\nThe idea behind this patch is to store traces as Redis lists under key equal\nto trace id. In this case list operation iterates only over unique trace ids\nand get operation retrieves content of specified list. Note that list operation\nstill needs to retrieve at least 1 event from the trace to get timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9d50e6b86e6ed30f26f8f8ce6bb68dd8839663d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9d50e6b86e6ed30f26f8f8ce6bb68dd8839663d2"}]},"branch":"refs/heads/master"},"9ee38b2a142588643bed04f753b325441dac4739":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-22 16:17:38.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/5"}}},"commit":{"parents":[{"commit":"dead4d653e8bb12c832354b6270747b8ae1dd4f2","subject":"Merge \"Don\u0027t fail if sqlalchemy driver fails to initialize\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/dead4d653e8bb12c832354b6270747b8ae1dd4f2"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-03-22 16:17:12.000000000","tz":60},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stores each tracing event under its own key,\nas result both list and get operations require full scan of the database.\nThe idea behind this patch is to store traces as Redis lists under key equal\nto trace id. In this case list operation iterates only over unique trace ids\nand get operation retrieves content of specified list. Note that list operation\nstill needs to retrieve at least 1 event from the trace to get timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9ee38b2a142588643bed04f753b325441dac4739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9ee38b2a142588643bed04f753b325441dac4739"}]},"branch":"refs/heads/master"},"46f722e7276c4f303a39017e5e11367ed24a2d11":{"kind":"REWORK","_number":6,"created":"2019-04-08 10:21:50.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/6"}}},"commit":{"parents":[{"commit":"dead4d653e8bb12c832354b6270747b8ae1dd4f2","subject":"Merge \"Don\u0027t fail if sqlalchemy driver fails to initialize\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/dead4d653e8bb12c832354b6270747b8ae1dd4f2"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-04-08 10:21:02.000000000","tz":120},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stored each tracing event under its own key,\nas result both list and get operations required a full scan of the database.\nWith this patch traces are stored as Redis lists under a key equal\nto trace id. So list operation iterates only over unique trace ids\nand get operation retrieves the content of a specified list. Note that\nlist operation still needs to retrieve at least 1 event from the trace\nin order to get a timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/46f722e7276c4f303a39017e5e11367ed24a2d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/46f722e7276c4f303a39017e5e11367ed24a2d11"}]},"branch":"refs/heads/master"},"147effea78e92ae341fd048b1c9c1ff54c0ced0f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-04-09 11:27:36.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/98/625198/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/625198/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/625198/7"}}},"commit":{"parents":[{"commit":"a9d7cb1842e6843dcd9f31afe9f300f9dcc5a357","subject":"Merge \"Allow OSPROFILER_TRACE_SQLALCHEMY to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a9d7cb1842e6843dcd9f31afe9f300f9dcc5a357"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-12-14 09:43:04.000000000","tz":60},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-04-09 11:27:18.000000000","tz":120},"subject":"Optimize storage schema for Redis driver","message":"Optimize storage schema for Redis driver\n\nThe original Redis driver stored each tracing event under its own key,\nas result both list and get operations required a full scan of the database.\nWith this patch traces are stored as Redis lists under a key equal\nto trace id. So list operation iterates only over unique trace ids\nand get operation retrieves the content of a specified list. Note that\nlist operation still needs to retrieve at least 1 event from the trace\nin order to get a timestamp.\n\nPerformance test is executed with 1000 traces each consisting 1000 events:\n * existing driver:\n   * write: 48 sec\n   * list:  41 sec\n   * get:   3.6 sec\n * optimized driver:\n   * write: 44 sec\n   * list:  4.1 sec\n   * get:   0.01 sec\n\nChange-Id: I09a122f91e3d26531965fe53d6df7c66f502de4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/147effea78e92ae341fd048b1c9c1ff54c0ced0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/147effea78e92ae341fd048b1c9c1ff54c0ced0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
