)]}'
{"id":"openstack%2Fosprofiler~630298","triplet_id":"openstack%2Fosprofiler~master~Id4f3a5c31167b2edae839f82a56d431900aba33e","project":"openstack/osprofiler","branch":"master","topic":"bug/1811260","hashtags":[],"change_id":"Id4f3a5c31167b2edae839f82a56d431900aba33e","subject":"Fix osprofiler usage in horizon","status":"ABANDONED","created":"2019-01-11 15:50:36.000000000","updated":"2020-08-06 15:15:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c5ff15c1bf515672aa15de53613c4ece390978a","_number":630298,"virtual_id_number":630298,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-09 09:32:41.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2019-09-09 09:31:56.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2019-09-09 11:20:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"value":-1,"date":"2019-09-09 09:31:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-08 12:37:53.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-02-15 11:36:16.000000000","updated_by":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"reviewer":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"state":"REVIEWER"},{"updated":"2019-02-16 08:09:58.000000000","updated_by":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"reviewer":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"state":"REVIEWER"},{"updated":"2019-09-09 09:32:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-09 11:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d22993bfdf7ca209f2fbcf6a66605119c56f85ec","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-01-11 15:50:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e0812409dba567b51cd8d1aaf04e69aa873be6c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-11 15:51:56.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/630298/1/306/ : SUCCESS\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"f6711b0a16996a1c51d15d84085a623fb1321a0f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-01-11 15:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a45ce031cdccd17bb7ad215bac8b09f83abb2e7c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-11 15:54:29.000000000","message":"Patch Set 2:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/630298/2/307/ : SUCCESS\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"e6b8ef6179d9c276e57ed2974f27234b39e3bfd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 17:39:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/630298/2/check/tempest-full/a6daaaf/ : SUCCESS in 1h 38m 10s\n- tempest-full-py3 http://logs.openstack.org/98/630298/2/check/tempest-full-py3/d091d24/ : SUCCESS in 1h 05m 32s\n- openstack-tox-cover http://logs.openstack.org/98/630298/2/check/openstack-tox-cover/b10bc0d/cover/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/630298/2/check/openstack-tox-lower-constraints/da8a243/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/98/630298/2/check/openstack-tox-pep8/b90d931/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/98/630298/2/check/openstack-tox-py27/c4e52f6/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/98/630298/2/check/openstack-tox-py35/39822bb/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/98/630298/2/check/openstack-tox-py36/5257d16/ : SUCCESS in 7m 25s\n- openstack-tox-docs http://logs.openstack.org/98/630298/2/check/openstack-tox-docs/e33c1e0/html/ : SUCCESS in 6m 33s\n- openstack-tox-functional http://logs.openstack.org/98/630298/2/check/openstack-tox-functional/c2a551e/ : SUCCESS in 3m 26s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/630298/2/check/openstack-tox-functional-py35/b0a7d1c/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3bb91f702530c23d0db8b3dee47d26aaea4b87b6","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-01-21 08:53:42.000000000","message":"Patch Set 2:\n\nMichal, I\u0027d like to try this out, but cannot find any information on how to enable profiling in Horizon. Do you have any docs on this?","accounts_in_message":[],"_revision_number":2},{"id":"c5b11fff28e165aef74337a0281bdc0762c4a9cd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-01-21 09:22:54.000000000","message":"Patch Set 2:\n\nIlya, sure , here is some older docs for ocata https://docs.openstack.org/horizon/ocata/topics/settings.html , but it is still actual. You have to just add this to local_settings.py in horizon \u003d\u003e   OPENSTACK_PROFILER \u003d { \u0027enabled\u0027 : False, \u0027keys\u0027: [\u0027secret\u0027], \u0027notifier_connection_string\u0027 : \u0027redis://127.0.0.1:6379\u0027, \u0027receiver_connection_string\u0027: \u0027redis://127.0.0.1:6379\u0027 }  , and you have to also copy or symlink files in openstack_dashboard/contrib/developer/enabled. Secret has to be defined also in osprofiler section of every openstack project , same as string for connection. If you will have some questions, let me know.","accounts_in_message":[],"_revision_number":2},{"id":"5edad9a98940e335e7c42be8bc96fe92874f305a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-01-21 09:41:21.000000000","message":"Patch Set 2:\n\nIf you will have some issues from horizon side , here is patch https://review.openstack.org/#/c/630372/","accounts_in_message":[],"_revision_number":2},{"id":"dcf6ee56303cd97c99fba29be516d560d2dae4a3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-01-21 13:54:14.000000000","message":"Patch Set 2:\n\nSorry, this https://review.openstack.org/#/c/631531/ and also this https://review.openstack.org/#/c/631531/","accounts_in_message":[],"_revision_number":2},{"id":"d69d5d7e58ac1983e9c276e1bf808f5c20ef04cc","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-02-15 11:36:16.000000000","message":"Patch Set 2: Code-Review-1\n\nThis patch breaks backward-compatibility as removes configuration parameters from the driver (e.g. profiler.socket_timeout). The change needs to be made on Horizon side to create CONF object and pass it to osprofiler. The similar approach was done to configure oslo.policy library.","accounts_in_message":[],"_revision_number":2},{"id":"f05c96140736286c6c856f436196392f8f91f337","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-02-15 11:59:55.000000000","message":"Patch Set 2:\n\nIlya, there is no backward compatibility , because the code which is currently in master never worked. People were just using mongodriver, and as you can see in mongodriver, they are not using oslo_config also.","accounts_in_message":[],"_revision_number":2},{"id":"385a87187929347a4c5c40c99eb28b9315a28f49","author":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"date":"2019-02-16 08:09:17.000000000","message":"Patch Set 2:\n\n\u003e Ilya, there is no backward compatibility , because the code which\n \u003e is currently in master never worked. People were just using\n \u003e mongodriver, and as you can see in mongodriver, they are not using\n \u003e oslo_config also.\n\njust mongodb driver? I don\u0027t think so. How about redis, elasticsearch?","accounts_in_message":[],"_revision_number":2},{"id":"c7b9d7c54dbf5693d414923babb8a81b208258b5","author":{"_account_id":23630,"name":"Tovin Seven","email":"tovin07@gmail.com","username":"tovin07"},"date":"2019-02-16 08:09:58.000000000","message":"Patch Set 2:\n\n\u003e Ilya, sure , here is some older docs for ocata https://docs.openstack.org/horizon/ocata/topics/settings.html\n \u003e , but it is still actual. You have to just add this to\n \u003e local_settings.py in horizon \u003d\u003e   OPENSTACK_PROFILER \u003d { \u0027enabled\u0027\n \u003e : False, \u0027keys\u0027: [\u0027secret\u0027], \u0027notifier_connection_string\u0027 :\n \u003e \u0027redis://127.0.0.1:6379\u0027, \u0027receiver_connection_string\u0027:\n \u003e \u0027redis://127.0.0.1:6379\u0027 }  , and you have to also copy or symlink\n \u003e files in openstack_dashboard/contrib/developer/enabled. Secret has\n \u003e to be defined also in osprofiler section of every openstack project\n \u003e , same as string for connection. If you will have some questions,\n \u003e let me know.\n\nDid you take a look at this patch?\n\nhttps://review.openstack.org/#/c/358574/","accounts_in_message":[],"_revision_number":2},{"id":"ae035baa9212358a91104c0b05d341602da16b36","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-02-23 13:13:28.000000000","message":"Patch Set 2:\n\n\u003e \u003e Ilya, there is no backward compatibility , because the code which\n \u003e \u003e is currently in master never worked. People were just using\n \u003e \u003e mongodriver, and as you can see in mongodriver, they are not\n \u003e using\n \u003e \u003e oslo_config also.\n \u003e \n \u003e just mongodb driver? I don\u0027t think so. How about redis,\n \u003e elasticsearch?\n\nElasticsearch and redis are implemented, yes, but I wanted to use redis, and it was not working. Why mongodbdriver init function using connection_str and other, let\u0027s say similar drivers using oslo_cfg which is not available in horizon and from my side of view neved worked . This is fixing this BUG, so i don\u0027t understand what is bad with this change ... Feel free to run devstack without a patch and run osprofiler in horizon with redis...and let me know ..","accounts_in_message":[],"_revision_number":2},{"id":"a082a4c1997710c2011b2928b8cd53a8367c3fe9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-02-23 13:16:35.000000000","message":"Patch Set 2:\n\n\u003e \u003e Ilya, sure , here is some older docs for ocata https://docs.openstack.org/horizon/ocata/topics/settings.html\n \u003e \u003e , but it is still actual. You have to just add this to\n \u003e \u003e local_settings.py in horizon \u003d\u003e   OPENSTACK_PROFILER \u003d {\n \u003e \u0027enabled\u0027\n \u003e \u003e : False, \u0027keys\u0027: [\u0027secret\u0027], \u0027notifier_connection_string\u0027 :\n \u003e \u003e \u0027redis://127.0.0.1:6379\u0027, \u0027receiver_connection_string\u0027:\n \u003e \u003e \u0027redis://127.0.0.1:6379\u0027 }  , and you have to also copy or\n \u003e symlink\n \u003e \u003e files in openstack_dashboard/contrib/developer/enabled. Secret\n \u003e has\n \u003e \u003e to be defined also in osprofiler section of every openstack\n \u003e project\n \u003e \u003e , same as string for connection. If you will have some questions,\n \u003e \u003e let me know.\n \u003e \n \u003e Did you take a look at this patch?\n \u003e \n \u003e https://review.openstack.org/#/c/358574/\n\nHi, yes, and what about it ?","accounts_in_message":[],"_revision_number":2},{"id":"05c0d691c5b2ddd21a55509be6e51e9f3b499b36","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-09-09 09:31:42.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8f70a853367ffdd74190a1a5dffe1e8776165341","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-09-09 09:31:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9fa1c992e84bedb994046dd57ec38dc2a7294d7b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-09 09:32:26.000000000","message":"Patch Set 3:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/630298/3/373/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"feb510ee63ce74befc804360981626f32d7106a9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-09 09:32:41.000000000","message":"Patch Set 3:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/630298/3/374/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"07d002c59315ac9913764cfdb090918d51e940d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 11:20:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c4f4ef8bb75e4e3f8746348b5364ef32 : SUCCESS in 1h 46m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/416a180200274940a5a9becdcc1860fa : SUCCESS in 1h 21m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f09cc958c6644c082fe52a94c60efed : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73a980ea414f47b6b593566a9c516ef7 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fb0102447b4e54a691d5806d3fc100 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef3291ca10ad454dabca4b8452031114 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed5df775cdb43f3953af28867b1e651 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53eaa2aedfc347b7b3a66a9b9332c583 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c65dcec0b4493a8df5607ccb1f4ba0 : SUCCESS in 2m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e0a81d2fbfda40e79a27d715dbb1a09b : SUCCESS in 3m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b124515c8c274a02b51759610b182c38 : SUCCESS in 5m 17s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/1a441c8e12b1471caae771ea29a57541 : SUCCESS in 48m 47s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/f414cc4bb73347bf99f75b513493493a : SUCCESS in 1h 07m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6167e41b5874d75d9121c1f8ae461c8198394651","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 15:15:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73399f3529f05e80f40abe049ed93e8239e5411d","revisions":{"6b51bb4901cde30d28a8b7c669e8127de2409c7c":{"kind":"REWORK","_number":1,"created":"2019-01-11 15:50:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/98/630298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/630298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/1"}}},"commit":{"parents":[{"commit":"97564b644f0446d03f8fab92632c84e9b2e43dec","subject":"Merge \"In case of an error, always add message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/97564b644f0446d03f8fab92632c84e9b2e43dec"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-01-11 15:38:21.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-01-11 15:50:32.000000000","tz":60},"subject":"Fix osprofiler usage in horizon","message":"Fix osprofiler usage in horizon\n\nThis patch fixes usage of osprofiler in horizon which is not\nworking now. Redis driver is not using oslo_config, and this patch\nis removing it as it removed in mongo driver already. As clients\nare calling osprofiler.initializer.init_from_conf, this patch\nis not affecting services. Only fixing osprofiler dashboard usage.\n\nChange-Id: Id4f3a5c31167b2edae839f82a56d431900aba33e\nCloses-Bug: #1811260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/6b51bb4901cde30d28a8b7c669e8127de2409c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/6b51bb4901cde30d28a8b7c669e8127de2409c7c"}]},"branch":"refs/heads/master"},"0f7118b7fb04bae05cfddd659f1a57b5a625cede":{"kind":"REWORK","_number":2,"created":"2019-01-11 15:53:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/98/630298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/630298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/2"}}},"commit":{"parents":[{"commit":"97564b644f0446d03f8fab92632c84e9b2e43dec","subject":"Merge \"In case of an error, always add message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/97564b644f0446d03f8fab92632c84e9b2e43dec"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-01-11 15:38:21.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-01-11 15:52:57.000000000","tz":60},"subject":"Fix osprofiler usage in horizon","message":"Fix osprofiler usage in horizon\n\nThis patch fixes usage of osprofiler in horizon which is not\nworking now. Redis driver is not using oslo_config, and this patch\nis removing it as it removed in mongo driver already. As clients\nare calling osprofiler.initializer.init_from_conf, this patch\nis not affecting services. Only fixing osprofiler dashboard usage.\n\nChange-Id: Id4f3a5c31167b2edae839f82a56d431900aba33e\nCloses-Bug: #1811260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0f7118b7fb04bae05cfddd659f1a57b5a625cede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0f7118b7fb04bae05cfddd659f1a57b5a625cede"}]},"branch":"refs/heads/master"},"73399f3529f05e80f40abe049ed93e8239e5411d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-09 09:31:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/98/630298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/98/630298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/98/630298/3"}}},"commit":{"parents":[{"commit":"d431c7a55153fc828ffa44dddd33d7d438e1ee3d","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/d431c7a55153fc828ffa44dddd33d7d438e1ee3d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-01-11 15:38:21.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-09-09 09:31:42.000000000","tz":0},"subject":"Fix osprofiler usage in horizon","message":"Fix osprofiler usage in horizon\n\nThis patch fixes usage of osprofiler in horizon which is not\nworking now. Redis driver is not using oslo_config, and this patch\nis removing it as it removed in mongo driver already. As clients\nare calling osprofiler.initializer.init_from_conf, this patch\nis not affecting services. Only fixing osprofiler dashboard usage.\n\nChange-Id: Id4f3a5c31167b2edae839f82a56d431900aba33e\nCloses-Bug: #1811260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/73399f3529f05e80f40abe049ed93e8239e5411d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/73399f3529f05e80f40abe049ed93e8239e5411d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
