)]}'
{"id":"openstack%2Fosprofiler~654114","triplet_id":"openstack%2Fosprofiler~master~Ifb379a40d90d252d56d4f830e75bbfa08bf66a01","project":"openstack/osprofiler","branch":"master","topic":"bug/1825693","hashtags":[],"change_id":"Ifb379a40d90d252d56d4f830e75bbfa08bf66a01","subject":"Minimum versions of databases with JSON data type support","status":"MERGED","created":"2019-04-21 04:29:15.000000000","updated":"2019-04-22 07:03:24.000000000","submitted":"2019-04-22 07:03:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654114-1555916604213-d1a6a7d1","meta_rev_id":"1161074672436121a18d4c9c5da28ddf62dcb088","_number":654114,"virtual_id_number":654114,"owner":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-04-22 07:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-22 06:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-22 06:55:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 06:55:09.000000000","updated_by":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"reviewer":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"state":"REVIEWER"},{"updated":"2019-04-22 07:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56dc31fa874ae329b2402304d49ce07dbd398e3c","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-04-21 04:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02248fe8b2968f56c61b9daf0fd7ee1d1546dce4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 04:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/654114/1/check/openstack-tox-pep8/20d1e74/ : SUCCESS in 2m 47s\n- openstack-tox-docs http://logs.openstack.org/14/654114/1/check/openstack-tox-docs/2dddaaa/html/ : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f13e161648f2a6a738ed868ccbb5d179d54cda6d","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-04-22 06:55:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"362113aa7bbe606b5c77543a5bcc3ce03c8c69b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 06:55:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2da7067738e92ecb6c3761b0e8a4c7952a91aff8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:03:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/654114/1/gate/openstack-tox-pep8/53bff74/ : SUCCESS in 3m 14s\n- openstack-tox-docs http://logs.openstack.org/14/654114/1/gate/openstack-tox-docs/8d17554/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1c27c4f6749feb27bed6f3e78f4613057953b258","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9847b8984a0e8ac9c325962431c08d7dddff13bc","revisions":{"9847b8984a0e8ac9c325962431c08d7dddff13bc":{"kind":"REWORK","_number":1,"created":"2019-04-21 04:29:15.000000000","uploader":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"ref":"refs/changes/14/654114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/14/654114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/14/654114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/14/654114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/14/654114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/14/654114/1"}}},"commit":{"parents":[{"commit":"7e76216ca1811821a697c1776a2b6dc7f6150070","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7e76216ca1811821a697c1776a2b6dc7f6150070"}]}],"author":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-04-21 04:22:43.000000000","tz":120},"committer":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-04-21 04:27:34.000000000","tz":120},"subject":"Minimum versions of databases with JSON data type support","message":"Minimum versions of databases with JSON data type support\n\nSQLAlchemy collector requires database JSON data type support\nwhich is fairly new feature and missing in versions older than:\n\n    - MariaDB 10.2\n    - MySQL 5.7.8\n\nLet\u0027s document that SQLAlchemy collector requirement.\n\nChange-Id: Ifb379a40d90d252d56d4f830e75bbfa08bf66a01\nCloses-Bug: #1825693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9847b8984a0e8ac9c325962431c08d7dddff13bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9847b8984a0e8ac9c325962431c08d7dddff13bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
