)]}'
{"id":"openstack%2Fosprofiler~659305","triplet_id":"openstack%2Fosprofiler~master~Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b","project":"openstack/osprofiler","branch":"master","hashtags":[],"change_id":"Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b","subject":"Automatic configuration of SQLAlchemy driver in DevStack","status":"MERGED","created":"2019-05-15 14:49:31.000000000","updated":"2019-05-20 09:25:35.000000000","submitted":"2019-05-20 09:25:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659305-1558344335642-16b933f0","meta_rev_id":"48be2e3db21f721c798d2a4e5641a2414bf3ca11","_number":659305,"virtual_id_number":659305,"owner":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-16 10:04:10.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":2,"date":"2019-05-20 09:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"recommended":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-16 13:46:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":2,"date":"2019-05-20 07:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-16 13:33:41.000000000","_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":1,"date":"2019-05-20 07:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 10:04:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-16 13:33:41.000000000","updated_by":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"reviewer":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"state":"REVIEWER"},{"updated":"2019-05-16 13:46:01.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-05-20 09:25:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"654035e6fef1633de93ad085406a331b7fb9c7ad","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-15 14:49:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed31d4ef715177220aa2dd45dc259829cd44955b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-15 14:49:57.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/1/357/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"b193601162642c2537f830d516773c19e8eeb481","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-15 14:51:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"838c28e5ea50a74d78c2947d52939bf21317a14c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-15 14:51:47.000000000","message":"Patch Set 2:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/2/358/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"690ce11a572e78a9f3aeb3914bab10b3a4a0506d","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-15 14:59:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d4e95793c2c0fd7cdf598162209c0d2c2e13554d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-15 15:00:17.000000000","message":"Patch Set 3:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/3/359/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"efd9f7a9596ab1a9c7840ba9088c1120910d3864","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/05/659305/3/check/tempest-full/c32f1d8/ : SUCCESS in 1h 52m 06s\n- tempest-full-py3 http://logs.openstack.org/05/659305/3/check/tempest-full-py3/02faf50/ : SUCCESS in 1h 29m 19s\n- openstack-tox-cover http://logs.openstack.org/05/659305/3/check/openstack-tox-cover/a2afb00/cover/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/659305/3/check/openstack-tox-lower-constraints/d007b81/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/05/659305/3/check/openstack-tox-pep8/7d004fa/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/05/659305/3/check/openstack-tox-py27/4772c54/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/05/659305/3/check/openstack-tox-py36/0a28320/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/05/659305/3/check/openstack-tox-py37/f72fc71/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/05/659305/3/check/openstack-tox-docs/726c9a1/html/ : SUCCESS in 3m 04s\n- openstack-tox-functional http://logs.openstack.org/05/659305/3/check/openstack-tox-functional/3f8bc62/ : SUCCESS in 2m 50s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/659305/3/check/openstack-tox-functional-py36/e07087d/ : SUCCESS in 3m 18s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/05/659305/3/check/tempest-smoke-py3-osprofiler-redis/81d639b/ : SUCCESS in 43m 24s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/05/659305/3/check/tempest-smoke-py3-osprofiler-sqlalchemy/26feff1/ : POST_FAILURE in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0436e8d4fdfc15459c1f02d3cc2db5df222c1449","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-15 20:09:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"66ca1e773f984b3e3e66b4b2c37c673f78512464","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-15 20:09:44.000000000","message":"Patch Set 4:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/4/360/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":4},{"id":"6e7af2e4640f03d967f34b7487da27394a0d1908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:06:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/05/659305/4/check/tempest-full/c56f65a/ : SUCCESS in 1h 51m 35s\n- tempest-full-py3 http://logs.openstack.org/05/659305/4/check/tempest-full-py3/37bf9b1/ : SUCCESS in 1h 50m 04s\n- openstack-tox-cover http://logs.openstack.org/05/659305/4/check/openstack-tox-cover/67341a7/cover/ : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/659305/4/check/openstack-tox-lower-constraints/fe19127/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/05/659305/4/check/openstack-tox-pep8/9106926/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/05/659305/4/check/openstack-tox-py27/db37af7/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/05/659305/4/check/openstack-tox-py36/22e8fc1/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/05/659305/4/check/openstack-tox-py37/09413b0/ : SUCCESS in 3m 49s\n- openstack-tox-docs http://logs.openstack.org/05/659305/4/check/openstack-tox-docs/3ac5ee9/html/ : SUCCESS in 2m 33s\n- openstack-tox-functional http://logs.openstack.org/05/659305/4/check/openstack-tox-functional/cf0c7cd/ : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/659305/4/check/openstack-tox-functional-py36/2837417/ : SUCCESS in 3m 10s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/05/659305/4/check/tempest-smoke-py3-osprofiler-redis/d165073/ : SUCCESS in 49m 38s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/05/659305/4/check/tempest-smoke-py3-osprofiler-sqlalchemy/36cdabe/ : POST_FAILURE in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9fb85603279ea785ac32073e93790b343943c71","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-16 07:49:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a2c91cc628ec9df77638dff12ad6fab2d2536a12","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-16 07:49:49.000000000","message":"Patch Set 5:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/5/361/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":5},{"id":"6c9f2ae5dbe987756521f63aa4a3317bf109ff9d","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-16 08:44:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"32199c78c5966486168ef3d22091b2611122b6a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-16 08:45:20.000000000","message":"Patch Set 6:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/6/362/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":6},{"id":"c334632fce9a3b1a1d25859d45c511696aab3f1d","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-16 10:03:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"58c8cb3c495194b5bf72236116b5792440c90b03","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-16 10:04:10.000000000","message":"Patch Set 7:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/659305/7/363/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":7},{"id":"e74572d364001581206afb76fc213dc27423268a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 12:01:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/05/659305/7/check/tempest-full/8a13e74/ : SUCCESS in 1h 50m 57s\n- tempest-full-py3 http://logs.openstack.org/05/659305/7/check/tempest-full-py3/7eb2bd6/ : SUCCESS in 1h 30m 33s\n- openstack-tox-cover http://logs.openstack.org/05/659305/7/check/openstack-tox-cover/985cf99/cover/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/659305/7/check/openstack-tox-lower-constraints/dd4d483/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/05/659305/7/check/openstack-tox-pep8/7be7574/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/05/659305/7/check/openstack-tox-py27/5cd12a5/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/05/659305/7/check/openstack-tox-py36/72bbb5c/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/05/659305/7/check/openstack-tox-py37/1d123e8/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/05/659305/7/check/openstack-tox-docs/1352907/html/ : SUCCESS in 3m 58s\n- openstack-tox-functional http://logs.openstack.org/05/659305/7/check/openstack-tox-functional/46311ea/ : SUCCESS in 3m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/659305/7/check/openstack-tox-functional-py36/05812b1/ : SUCCESS in 3m 02s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/05/659305/7/check/tempest-smoke-py3-osprofiler-redis/fed3e54/ : SUCCESS in 39m 04s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/05/659305/7/check/tempest-smoke-py3-osprofiler-sqlalchemy/178b630/ : SUCCESS in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"856a00edde156b1a243741e1c7220975f05f3d4f","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-05-16 13:46:01.000000000","message":"Patch Set 7: Code-Review+1\n\nCool. Thanks!","accounts_in_message":[],"_revision_number":7},{"id":"65154ee7d8af233ed6d9c11db6ba0054309d6737","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2019-05-20 07:32:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nReviewed by driver\u0027s author and verified by the integration job. Merging.","accounts_in_message":[],"_revision_number":7},{"id":"894f88c882409013f0ae9ccd8ffc7fe6e1c0c02d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:32:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7370b98800b85b38b2446fbc534461869342cc68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:25:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"48be2e3db21f721c798d2a4e5641a2414bf3ca11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:25:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/05/659305/7/gate/tempest-full/e07a671/ : SUCCESS in 1h 51m 46s\n- tempest-full-py3 http://logs.openstack.org/05/659305/7/gate/tempest-full-py3/023ba9a/ : SUCCESS in 1h 40m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/659305/7/gate/openstack-tox-lower-constraints/7dd5e4c/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/05/659305/7/gate/openstack-tox-pep8/4e9aa69/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/05/659305/7/gate/openstack-tox-py27/5bf1569/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/05/659305/7/gate/openstack-tox-py36/e3f6041/ : SUCCESS in 2m 37s\n- openstack-tox-py37 http://logs.openstack.org/05/659305/7/gate/openstack-tox-py37/052c5dc/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/05/659305/7/gate/openstack-tox-docs/743f0ea/html/ : SUCCESS in 2m 18s\n- openstack-tox-functional http://logs.openstack.org/05/659305/7/gate/openstack-tox-functional/b162ba1/ : SUCCESS in 3m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/659305/7/gate/openstack-tox-functional-py36/2e2186f/ : SUCCESS in 3m 25s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/05/659305/7/gate/tempest-smoke-py3-osprofiler-redis/15f8a41/ : SUCCESS in 42m 07s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/05/659305/7/gate/tempest-smoke-py3-osprofiler-sqlalchemy/3957630/ : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0fcc65d0d5a9df40dc53fb611a6058e18a44c654","revisions":{"0e33d0f5e2399b920cb1c6606a7cc107f36c018c":{"kind":"REWORK","_number":1,"created":"2019-05-15 14:49:31.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/1"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"subject":"[TEST] Configure sqlalchemy driver in devstack","message":"[TEST] Configure sqlalchemy driver in devstack\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0e33d0f5e2399b920cb1c6606a7cc107f36c018c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0e33d0f5e2399b920cb1c6606a7cc107f36c018c"}]},"branch":"refs/heads/master"},"da9638305bb30cf20650a99979ce311843fb3e15":{"kind":"REWORK","_number":2,"created":"2019-05-15 14:51:19.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/2"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:51:08.000000000","tz":120},"subject":"[TEST] Configure sqlalchemy driver in devstack","message":"[TEST] Configure sqlalchemy driver in devstack\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/da9638305bb30cf20650a99979ce311843fb3e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/da9638305bb30cf20650a99979ce311843fb3e15"}]},"branch":"refs/heads/master"},"750fdaee217f41d3d3a249b5d36a1d0278aa1189":{"kind":"REWORK","_number":3,"created":"2019-05-15 14:59:47.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/3"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:59:39.000000000","tz":120},"subject":"[TEST] Configure sqlalchemy driver in devstack","message":"[TEST] Configure sqlalchemy driver in devstack\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/750fdaee217f41d3d3a249b5d36a1d0278aa1189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/750fdaee217f41d3d3a249b5d36a1d0278aa1189"}]},"branch":"refs/heads/master"},"658707efe88f8564b801f1c1c378ec782c336e8c":{"kind":"REWORK","_number":4,"created":"2019-05-15 20:09:15.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/4"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 20:09:10.000000000","tz":0},"subject":"[TEST] Configure sqlalchemy driver in devstack","message":"[TEST] Configure sqlalchemy driver in devstack\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/658707efe88f8564b801f1c1c378ec782c336e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/658707efe88f8564b801f1c1c378ec782c336e8c"}]},"branch":"refs/heads/master"},"69baa671e32f0f502318521c07ef06e2a70f3d76":{"kind":"REWORK","_number":5,"created":"2019-05-16 07:49:18.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/5"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-16 07:45:33.000000000","tz":120},"subject":"Automatic configuration of SQLAlchemy driver in DevStack","message":"Automatic configuration of SQLAlchemy driver in DevStack\n\nThis patch adds a new type of collector \"sqlalchemy\".\nTo enable it, add the following line into DevStack local.conf:\n\n    OSPROFILER_COLLECTOR\u003dsqlalchemy\n\nOSProfiler DevStack plugin creates database for traces and\nconfigures all services to use it.\n\nZuul job is added to run Tempest tests with sqlalchemy driver enabled.\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/69baa671e32f0f502318521c07ef06e2a70f3d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/69baa671e32f0f502318521c07ef06e2a70f3d76"}]},"branch":"refs/heads/master"},"740db5933314c83664a65672d32f4ea4bee9f4d3":{"kind":"REWORK","_number":6,"created":"2019-05-16 08:44:54.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/6"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-16 08:44:42.000000000","tz":120},"subject":"Automatic configuration of SQLAlchemy driver in DevStack","message":"Automatic configuration of SQLAlchemy driver in DevStack\n\nThis patch adds a new type of collector \"sqlalchemy\".\nTo enable it, add the following line into DevStack local.conf:\n\n    OSPROFILER_COLLECTOR\u003dsqlalchemy\n\nOSProfiler DevStack plugin creates database for traces and\nconfigures all services to use it.\n\nZuul job is added to run Tempest tests with sqlalchemy driver enabled.\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/740db5933314c83664a65672d32f4ea4bee9f4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/740db5933314c83664a65672d32f4ea4bee9f4d3"}]},"branch":"refs/heads/master"},"0fcc65d0d5a9df40dc53fb611a6058e18a44c654":{"kind":"REWORK","_number":7,"created":"2019-05-16 10:03:41.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/05/659305/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/05/659305/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/05/659305/7"}}},"commit":{"parents":[{"commit":"30049d6c901e9a36f8e1dd07bac3bdeadd20b550","subject":"Merge \"Fix elasticsearch version in python requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30049d6c901e9a36f8e1dd07bac3bdeadd20b550"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-15 14:47:25.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2019-05-16 10:03:00.000000000","tz":120},"subject":"Automatic configuration of SQLAlchemy driver in DevStack","message":"Automatic configuration of SQLAlchemy driver in DevStack\n\nThis patch adds a new type of collector \"sqlalchemy\".\nTo enable it, add the following line into DevStack local.conf:\n\n    OSPROFILER_COLLECTOR\u003dsqlalchemy\n\nOSProfiler DevStack plugin creates database for traces and\nconfigures all services to use it.\n\nSQLAlchemy driver is extend to be able to list traces.\n\nZuul job is added to run Tempest tests with sqlalchemy driver enabled.\n\nChange-Id: Ia943d311d78f6dfd2e6bb884c8dca725b3b2a36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0fcc65d0d5a9df40dc53fb611a6058e18a44c654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0fcc65d0d5a9df40dc53fb611a6058e18a44c654"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
