)]}'
{"id":"openstack%2Fosprofiler~668730","triplet_id":"openstack%2Fosprofiler~master~I153668852ac414bdeaa00efaf7572154d42246b9","project":"openstack/osprofiler","branch":"master","topic":"f29","hashtags":[],"change_id":"I153668852ac414bdeaa00efaf7572154d42246b9","subject":"DNM: test patch to use f29 nodeset from devstack","status":"ABANDONED","created":"2019-07-02 18:27:21.000000000","updated":"2019-09-13 18:32:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"980433527ca5602be5f607342c2e389e9a157db9","_number":668730,"virtual_id_number":668730,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-02 18:30:25.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-07-02 20:14:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-02 18:27:31.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 18:30:25.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-02 20:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07d70ffcfb60318835bd849fcf2a01f41f9c1d28","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cd4574d6c9dc54232810c678c7833dd7c4ed6de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:27:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2b676afdca1654e9a8818c2e89aefe457013c4fe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-02 18:28:14.000000000","message":"Topic set to f29","accounts_in_message":[],"_revision_number":1},{"id":"a2f7a9e5cd9ff2cae6e4231b31a67cc9f3f5d5c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-02 18:30:25.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/668730/1/369/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"dcdcdb684e92396778ff051e443732fc39b1ab21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/30/668730/1/check/tempest-full/4930aa3/ : SUCCESS in 1h 43m 55s\n- tempest-full-py3 http://logs.openstack.org/30/668730/1/check/tempest-full-py3/ebabf48/ : SUCCESS in 1h 39m 28s\n- openstack-tox-cover http://logs.openstack.org/30/668730/1/check/openstack-tox-cover/1713fa3/cover/ : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/668730/1/check/openstack-tox-lower-constraints/06ff2b1/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/30/668730/1/check/openstack-tox-pep8/6ae832e/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/30/668730/1/check/openstack-tox-py27/c39eedc/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/30/668730/1/check/openstack-tox-py36/6a230a5/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/30/668730/1/check/openstack-tox-py37/1bbcc9b/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/30/668730/1/check/openstack-tox-docs/e0ded1f/html/ : SUCCESS in 2m 59s\n- openstack-tox-functional http://logs.openstack.org/30/668730/1/check/openstack-tox-functional/f7f7605/ : SUCCESS in 3m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/30/668730/1/check/openstack-tox-functional-py36/8e7a954/ : SUCCESS in 5m 26s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/30/668730/1/check/tempest-smoke-py3-osprofiler-redis/b2ed1c7/ : SUCCESS in 41m 10s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/30/668730/1/check/tempest-smoke-py3-osprofiler-sqlalchemy/8e6cec7/ : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"980433527ca5602be5f607342c2e389e9a157db9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-13 18:32:49.000000000","message":"Abandoned\n\nWas just a test patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc77c7929790f254adfa3a6b293079649ed8850d","revisions":{"bc77c7929790f254adfa3a6b293079649ed8850d":{"kind":"REWORK","_number":1,"created":"2019-07-02 18:27:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/30/668730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/30/668730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/30/668730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/30/668730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/30/668730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/30/668730/1"}}},"commit":{"parents":[{"commit":"ede0ec402eee45e107c1fde429c3332b3e49f60e","subject":"Bring env OSPROFILER_CONNECTION_STRING into effect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ede0ec402eee45e107c1fde429c3332b3e49f60e"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:27:04.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-02 18:27:20.000000000","tz":-240},"subject":"DNM: test patch to use f29 nodeset from devstack","message":"DNM: test patch to use f29 nodeset from devstack\n\nDepends-on: https://review.opendev.org/#/c/662529/\nChange-Id: I153668852ac414bdeaa00efaf7572154d42246b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/bc77c7929790f254adfa3a6b293079649ed8850d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/bc77c7929790f254adfa3a6b293079649ed8850d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
