)]}'
{"id":"openstack%2Fosprofiler~669466","triplet_id":"openstack%2Fosprofiler~master~I18d35d120bed22f5bbb4b8e15942d3e48063615d","project":"openstack/osprofiler","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I18d35d120bed22f5bbb4b8e15942d3e48063615d","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:12:53.000000000","updated":"2019-07-16 22:34:04.000000000","submitted":"2019-07-16 22:34:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669466-1563316444406-9977374c","meta_rev_id":"e74285fde1ea3430d73687b9bac023472fe89da4","_number":669466,"virtual_id_number":669466,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-05 19:15:59.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-16 22:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-16 20:30:33.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-16 20:49:13.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 19:15:59.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-16 20:49:13.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-16 22:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15ccbdd657ec226b970097048cf7065d5495979d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4864bc9833e34e57b0463c811232774ae4934e3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-05 19:15:59.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/669466/1/370/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"8465122ebfbecea417e6349eba603b90f9ddf001","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:02:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/66/669466/1/check/test-release-openstack/b5bd020/ : SUCCESS in 2m 49s\n- tempest-full http://logs.openstack.org/66/669466/1/check/tempest-full/8ca32eb/ : SUCCESS in 1h 32m 24s\n- tempest-full-py3 http://logs.openstack.org/66/669466/1/check/tempest-full-py3/040d066/ : SUCCESS in 1h 43m 45s\n- openstack-tox-cover http://logs.openstack.org/66/669466/1/check/openstack-tox-cover/ea53423/cover/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/669466/1/check/openstack-tox-lower-constraints/8a37b66/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/66/669466/1/check/openstack-tox-pep8/6d12d68/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/66/669466/1/check/openstack-tox-py27/b67074c/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/66/669466/1/check/openstack-tox-py36/03f52f0/ : SUCCESS in 3m 54s\n- openstack-tox-py37 http://logs.openstack.org/66/669466/1/check/openstack-tox-py37/1ea57d6/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/66/669466/1/check/openstack-tox-docs/06e2e73/html/ : SUCCESS in 2m 56s\n- build-openstack-releasenotes http://logs.openstack.org/66/669466/1/check/build-openstack-releasenotes/acf6bf1/html/ : SUCCESS in 3m 42s\n- openstack-tox-functional http://logs.openstack.org/66/669466/1/check/openstack-tox-functional/0b4c25e/ : SUCCESS in 3m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/66/669466/1/check/openstack-tox-functional-py36/bea7850/ : SUCCESS in 3m 53s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/66/669466/1/check/tempest-smoke-py3-osprofiler-redis/bf693cf/ : SUCCESS in 51m 40s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/66/669466/1/check/tempest-smoke-py3-osprofiler-sqlalchemy/939ba0e/ : SUCCESS in 51m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a9992dbe07bbffcd77c9cfb034bd431195b9813","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:30:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a066a3a61ad23d844ef9655ede1d6f17e4e9385f","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:49:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cab512c38fb0802be37d190a06ceab0a9a6e6be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:49:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22a6520d306beeac745e8a2a273d2a131618f482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 22:34:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/66/669466/1/gate/test-release-openstack/66f9408/ : SUCCESS in 3m 28s\n- tempest-full http://logs.openstack.org/66/669466/1/gate/tempest-full/1a74f90/ : SUCCESS in 1h 37m 36s\n- tempest-full-py3 http://logs.openstack.org/66/669466/1/gate/tempest-full-py3/b497b26/ : SUCCESS in 1h 31m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/669466/1/gate/openstack-tox-lower-constraints/dffb775/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/66/669466/1/gate/openstack-tox-pep8/fe9e4af/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/66/669466/1/gate/openstack-tox-py27/eb557d9/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/66/669466/1/gate/openstack-tox-py36/3523fa7/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/66/669466/1/gate/openstack-tox-py37/c5d2247/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/66/669466/1/gate/openstack-tox-docs/39d6bf5/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes http://logs.openstack.org/66/669466/1/gate/build-openstack-releasenotes/f6529ab/html/ : SUCCESS in 3m 03s\n- openstack-tox-functional http://logs.openstack.org/66/669466/1/gate/openstack-tox-functional/fcd28a4/ : SUCCESS in 3m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/66/669466/1/gate/openstack-tox-functional-py36/012b263/ : SUCCESS in 4m 43s\n- tempest-smoke-py3-osprofiler-redis http://logs.openstack.org/66/669466/1/gate/tempest-smoke-py3-osprofiler-redis/6bbbac5/ : SUCCESS in 42m 15s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy http://logs.openstack.org/66/669466/1/gate/tempest-smoke-py3-osprofiler-sqlalchemy/4eefa92/ : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efcb636f68018133c7a6b1968eff80063c4e9d22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 22:34:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d431c7a55153fc828ffa44dddd33d7d438e1ee3d","revisions":{"d431c7a55153fc828ffa44dddd33d7d438e1ee3d":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:12:53.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/66/669466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/66/669466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/66/669466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/66/669466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/66/669466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/66/669466/1"}}},"commit":{"parents":[{"commit":"ede0ec402eee45e107c1fde429c3332b3e49f60e","subject":"Bring env OSPROFILER_CONNECTION_STRING into effect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ede0ec402eee45e107c1fde429c3332b3e49f60e"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:58.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:12:53.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I18d35d120bed22f5bbb4b8e15942d3e48063615d\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/d431c7a55153fc828ffa44dddd33d7d438e1ee3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/d431c7a55153fc828ffa44dddd33d7d438e1ee3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
