)]}'
{"id":"openstack%2Fosprofiler~672724","triplet_id":"openstack%2Fosprofiler~stable%2Focata~Ife6bcaca347d914cb1fc7e2fe71665ea5c93ca80","project":"openstack/osprofiler","branch":"stable/ocata","hashtags":[],"change_id":"Ife6bcaca347d914cb1fc7e2fe71665ea5c93ca80","subject":"Use upper constraints","status":"MERGED","created":"2019-07-25 13:44:16.000000000","updated":"2019-07-25 18:41:01.000000000","submitted":"2019-07-25 18:41:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672724-1564080061318-cd2898e8","meta_rev_id":"0f5e40d93755b690d2cbe6d19b3af2e27e3a1ef8","_number":672724,"virtual_id_number":672724,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-25 18:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-07-25 15:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-07-25 15:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-07-25 15:33:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 15:28:46.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-25 15:33:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-25 18:41:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c846ed3f53580bce513121760e9482e951c162f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-25 13:44:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"137eec48a3f891ccc04f2cb76fa6b05289f4fbc3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-25 14:09:10.000000000","message":"Patch Set 1: Code-Review+2\n\nNeed to wait for CI results of course but this looks sensible so straight up +2. Thanks for following through on all of these, Elod.","accounts_in_message":[],"_revision_number":1},{"id":"e742496b6dbe1c6406e92e04aa83bfb549d0a027","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-25 14:10:22.000000000","message":"Patch Set 1:\n\nThis patch fixes Periodic Stable job, which currently fails. See:\n\nhttp://logs.openstack.org/periodic-stable/opendev.org/openstack/osprofiler/stable/ocata/openstack-tox-py27/102ed21/job-output.txt.gz#_2019-07-25_06_47_57_119626","accounts_in_message":[],"_revision_number":1},{"id":"b184cab233d36f76bfc262c54b1824df0c31d9eb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-07-25 14:12:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"748f5c0144f1b06969330959ee80d582c067441d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-25 15:28:46.000000000","message":"Patch Set 2: Code-Review+2\n\nAgreed that this seems fine.","accounts_in_message":[],"_revision_number":2},{"id":"0d018abb2c271fe808023ae8b9a21450a5a2cccd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:32:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/24/672724/2/check/openstack-tox-pep8/8ff0f06/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/24/672724/2/check/openstack-tox-py27/9e9c4d8/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/24/672724/2/check/openstack-tox-py35/454db86/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/24/672724/2/check/openstack-tox-docs/206db06/html/ : SUCCESS in 3m 07s\n- build-openstack-releasenotes http://logs.openstack.org/24/672724/2/check/build-openstack-releasenotes/9b7f449/html/ : SUCCESS in 3m 20s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/24/672724/2/check/legacy-tempest-dsvm-neutron-src/2a29744/ : FAILURE in 1h 12m 29s\n- openstack-tox-cover http://logs.openstack.org/24/672724/2/check/openstack-tox-cover/b94184b/cover/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1e180e15feed1e6a80f8f9300f9e1a3999c30e43","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-25 15:33:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":2},{"id":"3179a001cf8f05553e2912be22d9a3f5b214154b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 17:24:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/24/672724/2/check/openstack-tox-pep8/f749934/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/24/672724/2/check/openstack-tox-py27/3055c08/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/24/672724/2/check/openstack-tox-py35/1d8ad35/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/24/672724/2/check/openstack-tox-docs/7de0694/html/ : SUCCESS in 3m 15s\n- build-openstack-releasenotes http://logs.openstack.org/24/672724/2/check/build-openstack-releasenotes/6b7d691/html/ : SUCCESS in 3m 26s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/24/672724/2/check/legacy-tempest-dsvm-neutron-src/db80a73/ : SUCCESS in 1h 38m 43s\n- openstack-tox-cover http://logs.openstack.org/24/672724/2/check/openstack-tox-cover/ce0f036/cover/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f655667602bea35ee108a0ca0eba383718d69c9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 17:24:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18c36db6821edf0bacf7d30c8a96a3219caf2602","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:41:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/24/672724/2/gate/openstack-tox-pep8/b7eff5d/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/24/672724/2/gate/openstack-tox-py27/7d70b30/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/24/672724/2/gate/openstack-tox-py35/4f7791d/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/24/672724/2/gate/openstack-tox-docs/2c0b46c/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes http://logs.openstack.org/24/672724/2/gate/build-openstack-releasenotes/61500a4/html/ : SUCCESS in 4m 56s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/24/672724/2/gate/legacy-tempest-dsvm-neutron-src/624118b/ : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0f5e40d93755b690d2cbe6d19b3af2e27e3a1ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:41:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7858fbf43be1407fd22ff6d7941e0ed5399a1489","revisions":{"4bcdcba76cae45c2aa7e3d76fc5a7553bda009c2":{"kind":"REWORK","_number":1,"created":"2019-07-25 13:44:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/24/672724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/24/672724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/1"}}},"commit":{"parents":[{"commit":"eefe506a72da1c154a09b5da2830cf5a690d8213","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/eefe506a72da1c154a09b5da2830cf5a690d8213"}]}],"author":{"name":"Elod Illes","email":"elod.illes@ericsson.com","date":"2019-05-10 13:33:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-25 13:41:59.000000000","tz":120},"subject":"Use upper constraints","message":"Use upper constraints\n\nSphinx 2.0.0 dropped python 2.7 and 3.4 support. Because osprofiler\nuses global upper-constraints.txt only from Rocky,\njob openstack-tox-py27 tries to install Sphinx 2.0.1 and fails [1] with:\n\n Downloading http://mirror.iad.rax.openstack.org/wheel/ubuntu-16.04-x86_64/sphinx/Sphinx-2.0.1-py2.py3-none-any.whl (3.2MB)\n ERROR: Sphinx requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.12\n\nNote: .[oslo.config] could be removed as the packages will be installed\nperfectly and constrained without this and it breaks installation if\nupper-constraints.txt is used anyway.\n\nNote: patch adjusted to Ocata, and in Ocata this is needed due to\nGitPython as it dropped support for python 2.7\n\n[1] http://logs.openstack.org/periodic-stable/opendev.org/openstack/osprofiler/stable/queens/openstack-tox-py27/3d469d3/job-output.txt.gz#_2019-05-10_06_10_56_192617\n\nChange-Id: Ife6bcaca347d914cb1fc7e2fe71665ea5c93ca80\n(cherry picked from commit eff57e09b55aaf1f00b01c289526ceddefdf25d9)\n(cherry picked from commit bfe78732b9991f11986099cd9cfa1b5afd94e0a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4bcdcba76cae45c2aa7e3d76fc5a7553bda009c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4bcdcba76cae45c2aa7e3d76fc5a7553bda009c2"}]},"branch":"refs/heads/stable/ocata"},"7858fbf43be1407fd22ff6d7941e0ed5399a1489":{"kind":"REWORK","_number":2,"created":"2019-07-25 14:12:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/24/672724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/24/672724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/24/672724/2"}}},"commit":{"parents":[{"commit":"eefe506a72da1c154a09b5da2830cf5a690d8213","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/eefe506a72da1c154a09b5da2830cf5a690d8213"}]}],"author":{"name":"Elod Illes","email":"elod.illes@ericsson.com","date":"2019-05-10 13:33:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-07-25 14:12:10.000000000","tz":120},"subject":"Use upper constraints","message":"Use upper constraints\n\nSphinx 2.0.0 dropped python 2.7 and 3.4 support. Because osprofiler\nuses global upper-constraints.txt only from Rocky,\njob openstack-tox-py27 tries to install Sphinx 2.0.1 and fails [1] with:\n\n Downloading http://mirror.iad.rax.openstack.org/wheel/ubuntu-16.04-x86_64/sphinx/Sphinx-2.0.1-py2.py3-none-any.whl (3.2MB)\n ERROR: Sphinx requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.12\n\nNote: .[oslo.config] could be removed as the packages will be installed\nperfectly and constrained without this and it breaks installation if\nupper-constraints.txt is used anyway.\n\nNote: patch adjusted to Ocata, and in Ocata this is needed due to\nGitPython as it dropped support for python 2.7\n\n[1] http://logs.openstack.org/periodic-stable/opendev.org/openstack/osprofiler/stable/queens/openstack-tox-py27/3d469d3/job-output.txt.gz#_2019-05-10_06_10_56_192617\n\nChange-Id: Ife6bcaca347d914cb1fc7e2fe71665ea5c93ca80\n(cherry picked from commit eff57e09b55aaf1f00b01c289526ceddefdf25d9)\n(cherry picked from commit bfe78732b9991f11986099cd9cfa1b5afd94e0a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7858fbf43be1407fd22ff6d7941e0ed5399a1489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7858fbf43be1407fd22ff6d7941e0ed5399a1489"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
