)]}'
{"id":"openstack%2Fosprofiler~705524","triplet_id":"openstack%2Fosprofiler~master~I2cc764bd3de8130aa25868b57b2e89341c71617e","project":"openstack/osprofiler","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I2cc764bd3de8130aa25868b57b2e89341c71617e","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2020-02-03 19:19:02.000000000","updated":"2020-02-04 14:36:19.000000000","submitted":"2020-02-04 14:30:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705524-1580826656723-5f1f21a0","meta_rev_id":"df04cdea161b392964ebdbc5da879208ac01c68a","_number":705524,"virtual_id_number":705524,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-04 10:24:52.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 14:30:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-04 13:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-02-04 10:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-02-04 13:03:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 08:18:35.000000000","updated_by":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"reviewer":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"state":"REVIEWER"},{"updated":"2020-02-04 10:24:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-04 10:30:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-04 13:03:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-04 14:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a28a15ca35755f3fabdb95e42009577480700053","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-03 19:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e200916da7da46c505dc1f3c916a42cb3272581","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-03 19:20:33.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/705524/1/380/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"e3f3dcc3b4edf748c05c829fcb805d9c2a2de701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 20:40:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efa936cbaf9c4471a148e23277764980 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4756fffb146f44dbbe9baa7870157cfd : SUCCESS in 3m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27e85dce30394af2a92c29539cc9c207 : SUCCESS in 1h 18m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12ccad5a58f741cdb235befe9820a433 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29b291b177fc44128bc4f97c07f7b250 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6ecfbe9dea41588b8e094ec144f9f7 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/134020f4ec9f43d7bb5df835d3e3045b : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc3d2510defc4ff28256ccb731ae23fc : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782a85b60b7f4790b9bf76b7326ad68d : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac439ea106e349d08a89b245ef7345a3 : FAILURE in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee55edd01774a939217b92bbf08051c : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08a7d13c7e794f0da95e20a9c1292152 : SUCCESS in 6m 26s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/55190b57af794687b7f0402d1a457a2b : SUCCESS in 40m 53s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/1bc1c5d87bfc410db28de28e5fab8eaf : SUCCESS in 57m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b6121d7f3de7484ceabff17a1b5c9f13b3f8f78","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-03 20:51:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad4ee1c0bac9c52b93162bc308a709f387d710b4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-03 20:59:19.000000000","message":"Patch Set 2:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/705524/2/381/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"bbc16d077438a3ca38de3e8c1f7d9bbcbc901a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 22:58:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b7fbc6900166402b8ef86a27e5500c57 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0ae84b9e11c4eab87f5791ddc8ec425 : SUCCESS in 5m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc8d6eeb5feb48cb8f749c9cb18018ad : SUCCESS in 2h 04m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5597b70313b4717b558959540edc8bc : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b895a97c306d4d509881fb768d1827f2 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af91040d4b5e48108faa17cd091f0696 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7200a804500421ab9a7bb6d36aa7bd2 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0a5721193554ebe9146b11f073a6830 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5e38aa054144fa196609656f7c3d73f : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce2e2b7457b4e56b6bcd1c2593aaaed : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/856368acaf1b41059728317c51a07c73 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ebbabc147e04572bcb34e63cda07604 : SUCCESS in 7m 12s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/bc8e5c1d17454e62a29b8fe6dc7ddbb9 : SUCCESS in 42m 36s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/ff01f03c001140edb6daf55b4712ad69 : SUCCESS in 42m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31ef003025c1ef66d2fe54b18652a7019f6c2bd3","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2020-02-04 08:18:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4dc0c2cd8ea630f596d0384bfce294582695fe91","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 09:41:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nI think the functional tests need to change","accounts_in_message":[],"_revision_number":2},{"id":"3b893c01314d81f1a28e6bdd98bd559c4d807d34","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 09:45:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6320486a70eefb9f157e7de9d0eaf751f557f8c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 10:04:32.000000000","message":"Patch Set 2: Code-Review-1\n\nplease drop this too \u003d\u003e https://github.com/openstack/osprofiler/blob/master/setup.py#L17,L23","accounts_in_message":[],"_revision_number":2},{"id":"89feff42d5c3b538e4eb0e471c369fe6055436da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 10:23:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9e5172bd39c23b4d2506b25a0c5dbafbb214c15","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-04 10:24:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ca3315332c913a18ca93b96e8e7fa2547786d2c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-04 10:24:52.000000000","message":"Patch Set 3:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/705524/3/382/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"dc4a843c5360b3bfdd066eead09bed60631d51e9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-04 10:30:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"96ea0b56128d63b6a2954d187ad8165ab4310276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 11:57:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b6f1381f7cc4ac2a8ed9bafacc44569 : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6598f1baa0304171bd8f49338130a800 : SUCCESS in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4174d0cb57484e95b7aaa8c6c4d76cc9 : SUCCESS in 1h 28m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8da5b61fa91644e8bae07aa028965408 : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aeed6be66f34f578244136596d1737b : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11406a7564894d999fe8f2835b97237a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e55618413a44cc59a10da786176783e : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f5dcd1ca13441a29c31f0a4edf31a8d : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bcab25d554f4dcd9de43868fba3be4d : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581ceedf415d4ef2bcbec5423010be31 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/075e221bade949ccb9faf6c374c578ce : SUCCESS in 4m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b2fd8e655ff44a089752f859fc03ae0f : SUCCESS in 3m 54s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/666a97ebfb454d56b5c001bbaa2d603e : SUCCESS in 41m 04s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/7b2c1a6292d64a42a340733417bf861a : SUCCESS in 41m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ed3128d87946d553e5a01962fe452963c267f55","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 13:02:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"329b990146969e25640b8740069672b69493e688","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-04 13:03:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"767ffb07c2ac3975d87d7c7c5c2c67f37307ecb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:03:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42cd7509ca1cfeed347d348029210bb28fdaf2f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 14:30:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fdc5553ceb03788a89b43e02ede03ea353feb4a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 14:30:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b1320d08a9b4e45aa26bf2d8208ad2c : SUCCESS in 3m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec3595e842e74a7ca0eaa1fa27170538 : SUCCESS in 4m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccac98cfe8774a4192796d3b0d604ee6 : SUCCESS in 1h 21m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df5578e22fd441f580e699d1901f1ffd : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67af38539cd44beb0e39ef1a7b30a34 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1066bad6091d48fca52a54349e297795 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/608699553149455b8bdc8712f800a81a : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f1c7eb1ff64fc2bac0b7c786006339 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c476cffda434071ac7bdb319c6fcd01 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2568e9e8b52a4fc785b58c2944e0cbf9 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fd61603386eb0d4e17f84350f659871cd25411dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 14:36:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a2ddcaaf9143c6af01d092811604f1 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a3caa33d7242d4bde7895634742ded : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7eaa8648cd4965a2fd5826cb064c8372f34a95c8","revisions":{"a12ab8f4c2156385efd551a10bb7c42581744670":{"kind":"REWORK","_number":1,"created":"2020-02-03 19:19:02.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/705524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/24/705524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/1"}}},"commit":{"parents":[{"commit":"e022fdaa8e6c69d4b43ebba96d70936bea18db77","subject":"Merge \"Handle driver initialization errors to avoid service crash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e022fdaa8e6c69d4b43ebba96d70936bea18db77"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 19:05:16.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 19:18:55.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch to using modern sphinx-build.\nRemove cruft from setup.cfg\n\nDrop non-voting jobs from gate queue.\n\nChange-Id: I2cc764bd3de8130aa25868b57b2e89341c71617e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a12ab8f4c2156385efd551a10bb7c42581744670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a12ab8f4c2156385efd551a10bb7c42581744670"}]},"branch":"refs/heads/master"},"72612a62f557e079d7dbfc2172a9df7173182324":{"kind":"REWORK","_number":2,"created":"2020-02-03 20:51:46.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/705524/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/24/705524/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/2"}}},"commit":{"parents":[{"commit":"e022fdaa8e6c69d4b43ebba96d70936bea18db77","subject":"Merge \"Handle driver initialization errors to avoid service crash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e022fdaa8e6c69d4b43ebba96d70936bea18db77"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 19:05:16.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 20:51:40.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch to using modern sphinx-build.\nRemove cruft from setup.cfg\n\nDrop non-voting jobs from gate queue.\n\nChange-Id: I2cc764bd3de8130aa25868b57b2e89341c71617e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/72612a62f557e079d7dbfc2172a9df7173182324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/72612a62f557e079d7dbfc2172a9df7173182324"}]},"branch":"refs/heads/master"},"7eaa8648cd4965a2fd5826cb064c8372f34a95c8":{"kind":"REWORK","_number":3,"created":"2020-02-04 10:24:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/24/705524/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/24/705524/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/24/705524/3"}}},"commit":{"parents":[{"commit":"e022fdaa8e6c69d4b43ebba96d70936bea18db77","subject":"Merge \"Handle driver initialization errors to avoid service crash\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e022fdaa8e6c69d4b43ebba96d70936bea18db77"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-03 19:05:16.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-04 10:24:14.000000000","tz":60},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nSwitch to using modern sphinx-build.\nRemove cruft from setup.cfg\n\nDrop non-voting jobs from gate queue.\n\nChange-Id: I2cc764bd3de8130aa25868b57b2e89341c71617e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7eaa8648cd4965a2fd5826cb064c8372f34a95c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7eaa8648cd4965a2fd5826cb064c8372f34a95c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
