)]}'
{"id":"openstack%2Fosprofiler~721161","triplet_id":"openstack%2Fosprofiler~master~I09e9c6970147ce1fe2880f1a2716268f69039c44","project":"openstack/osprofiler","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I09e9c6970147ce1fe2880f1a2716268f69039c44","subject":"Use unittest.mock instead of third party mock","status":"ABANDONED","created":"2020-04-20 03:26:30.000000000","updated":"2020-04-20 09:08:31.000000000","total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6a0e5ed2398ea2b6ae7a002b774f037010a2301","_number":721161,"virtual_id_number":721161,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-20 06:05:59.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-20 07:29:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 06:05:59.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-20 07:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"530c84f450d11bcc90a0db074d30f8d5a0b5dd28","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 03:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cdc798012986fc17c5474a83ad967ba3452156e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-20 03:27:02.000000000","message":"Patch Set 1:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/721161/1/388/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":1},{"id":"b9081e7e003cb3826fb0b58108b48937218440dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 04:47:34.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d678256afee7457ca60a68450160be6c : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97d50526150a4f8381053809e219d838 : SUCCESS in 1h 16m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d132b20487b48b29d233f17000eed8b : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ec84ced812b4432bf73ad270a77aefe : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aefdb7005466414197e41fd066fda3b6 : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7322c63c984844a44830e8109d734c : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49963ef8019644b99648ea7ac75df0c6 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e1fb89929845b2a9d6eba8414dcb92 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8798c98c33bf42a8ab54962e7b173aa9 : SUCCESS in 3m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7e23f8d280842969cc8ff96f27d23ab : SUCCESS in 3m 33s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/bbb9e2357b804b7d959c76f72353e7f3 : SUCCESS in 37m 54s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/09a868144644408e91b07e95313d0162 : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa3ed9a1c225a60d54adf9da41775b838ffd13cc","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 06:05:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"290ce034e6133ca9f4614b73a02a6b9ddee4fe09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-20 06:05:59.000000000","message":"Patch Set 2:\n\n- ext-osprofiler-functional-LogInsight http://207.189.188.190/logs/ext-osprofiler-functional-LogInsight/721161/2/389/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"d1fd57189e5c63da26a15cb3900bfa4d163a8f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 07:29:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/262be1aaee7a4d1cbd86301e6f426332 : SUCCESS in 5m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35fad9239a0547a58cdc3c378d50339f : SUCCESS in 1h 14m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7713b3b5d2d84f46ac348e47ee27d445 : SUCCESS in 8m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09dfb0e8f18d4ba78afdd490e3956acd : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86edf55f8894876a39a367bb7b12b6f : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b0344c069c4e6589f05f5a318762ec : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbcf3d2047de48c98db90ba5c600a8fe : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0b5d00d96c459981dd71897abd2876 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39e7afa99cf8456d9ff0474a0a26a920 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/906dfa2bae5f4d508e21406a9c5a16b3 : SUCCESS in 7m 24s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/f40d711efdb9480a91198e81d0c839b8 : SUCCESS in 40m 09s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/3817efa90ed644aab266d5b67b0c4bd4 : FAILURE in 44m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b856b0dc9ca1f450ecf51777ea2829c0a72aaba","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 08:05:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"74034dd6f1f33b627a4d36048407a1a9060a71c1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-20 08:58:00.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6356fde2f8fd3a76546765e77ac0935b1ced6bd5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-20 08:58:22.000000000","message":"Patch Set 2:\n\nwhoops, didn\u0027t notice this was abandoned /o\\","accounts_in_message":[],"_revision_number":2},{"id":"140c213015ce6238caf962ba6076ae71c7f02366","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 09:08:31.000000000","message":"Patch Set 2:\n\n\u003e whoops, didn\u0027t notice this was abandoned /o\\\n\nthanks Stephen. I didn\u0027t notice this was committed, so abandoned this patch after notice","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca9bfd8d5572134ea6117a6227f57deb81238c9c","revisions":{"669483cea00ca36581c26b457eee9cb0b87c1402":{"kind":"REWORK","_number":1,"created":"2020-04-20 03:26:30.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/61/721161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/61/721161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/1"}}},"commit":{"parents":[{"commit":"3c95bd58a008257ff6a0b030155945fac9792d6c","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/3c95bd58a008257ff6a0b030155945fac9792d6c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 03:23:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 03:26:27.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I09e9c6970147ce1fe2880f1a2716268f69039c44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/669483cea00ca36581c26b457eee9cb0b87c1402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/669483cea00ca36581c26b457eee9cb0b87c1402"}]},"branch":"refs/heads/master"},"ca9bfd8d5572134ea6117a6227f57deb81238c9c":{"kind":"REWORK","_number":2,"created":"2020-04-20 06:05:19.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/61/721161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/61/721161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/61/721161/2"}}},"commit":{"parents":[{"commit":"3c95bd58a008257ff6a0b030155945fac9792d6c","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/3c95bd58a008257ff6a0b030155945fac9792d6c"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 03:23:47.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 06:05:02.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I09e9c6970147ce1fe2880f1a2716268f69039c44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ca9bfd8d5572134ea6117a6227f57deb81238c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ca9bfd8d5572134ea6117a6227f57deb81238c9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
