)]}'
{"id":"openstack%2Fosprofiler~739379","triplet_id":"openstack%2Fosprofiler~master~Ib1870a3d102116f84c7677601fd44fdac41a13a6","project":"openstack/osprofiler","branch":"master","topic":"osc-performance","hashtags":[],"change_id":"Ib1870a3d102116f84c7677601fd44fdac41a13a6","subject":"switch to importlib.metadata to find package version","status":"MERGED","created":"2020-07-05 23:35:52.000000000","updated":"2020-07-21 00:41:54.000000000","submitted":"2020-07-21 00:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739379-1595291670275-c1f85032","meta_rev_id":"61db336aed62e983051622b09b08534904df1cee","_number":739379,"virtual_id_number":739379,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 00:34:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-07-20 15:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-06 19:03:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-07-20 15:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2020-07-06 09:07:19.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 09:07:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-06 19:03:43.000000000","updated_by":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"reviewer":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"state":"REVIEWER"},{"updated":"2020-07-20 15:32:04.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-21 00:34:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4675c7c19bee7cbc85d985645450d7f49abd3dcf","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2020-07-05 23:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"125ecec61a0cde115f12404d3b4aed264c594c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 00:54:23.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6310c71988fb45d5b24b7fc85fe02555 : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa507da7d47144c4be3e2e3ca5025436 : SUCCESS in 1h 13m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce0e8e9ca1c841d9a0c5261aeb56465a : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddc539a2fd944021935a12a705418002 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b47c5b01dba4644becc97a05ff3974e : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d09a5749aa147bfa5d29f27e2d92c07 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e569adbd88f5404ba330f24e4f736072 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcadf12a934443ab84fcaac4cf2217c : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17fc6808d99842aeb7e2f77efad35ca4 : SUCCESS in 5m 01s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/f5ffe4f09ab84d9ca6a27177b3a1c2ca : SUCCESS in 40m 34s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/e6aef60b7afa45a1875acfabd91a64ad : SUCCESS in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"229c60aa4d54d4322e26be6fe6711b30ea0af288","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-06 09:07:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54db33874512f71c482b46c409bb90af1ec4960e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 10:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/29631e21c9f54fa78791c6cb688c4fd8 : SUCCESS in 4m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97a91a2c70694396adadf27e457eee78 : SUCCESS in 1h 36m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daebf025ae8043148a4333298140a572 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0bf43b1efc94258959cb5035959df6c : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbfeb7fc61a431d8d6651e18a400c05 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a112c49254c640c587230cd89fc58830 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bc51d6fabc14021a52848b9b698b62b : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb48003d24424ac7b6740f9ba0422687 : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/668a1b7686224bce8bae33de520a99b1 : SUCCESS in 5m 16s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/a61cd35899b54e97a387ced94a278cb2 : SUCCESS in 34m 03s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/d42e3e1d927541cfba7a65349000cdb7 : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1cb4e30532f7060cf19318ad88b6a96bc368d239","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2020-07-06 19:03:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3228657619510a7d5bbb3deb3406a3a9d984ec52","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-20 15:32:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56bfd6861a6a6e71fd4f63eec6ad5459160bdcf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:32:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f948874912a99dcbec8b91f2d397296763754f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:06:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/57e3dd65aa0d46b4a4de5623c2148d70 : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5480f8197514402babf45c15122ab396 : FAILURE in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1704d4010d843499683f91ac1c6a5f4 : SUCCESS in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225b6a3cb87349f6ac1b020db70bca4f : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15aef5d55d1b48a38c200f3992bca96f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a76091b4c7b4053b86eeb053491758d : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950607608505422d82c627f5a170b7f3 : SUCCESS in 4m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a685051238344089110f58e89311b59 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8e76af317c2e535b4170fdf0838d8090c0286757","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-20 22:11:17.000000000","message":"Patch Set 2:\n\nrecheck pip failure","accounts_in_message":[],"_revision_number":2},{"id":"18022426ee393fc93c4835f1bb39391d78a69130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:25:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/801c2a2f562a4703a3978a288966f788 : SUCCESS in 3m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abe5d3f1901841cda7a5ac89baa3252c : SUCCESS in 1h 13m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68dc54db66d340d589b865c18b1908af : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf9f9faa379e4fc48e2a9fb902f31493 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0516690b5c649008452b93464a02fc1 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bfeb4e25b4b49a4bf57baca310c1780 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5ffe3fe1398442782003b0ae7431120 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a103e0454f4341d6b8a3ece5823431d1 : SUCCESS in 3m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bffa16ee2634a8c9c93e74ff64bcb38 : SUCCESS in 5m 43s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/56b7364e298f49dabed284195d82d072 : SUCCESS in 44m 07s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/78652d60434f43dabe1fcd2149df6a6c : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92ac5457f035c13c70971eaeb724966c5ff2d3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:25:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80ac149dcf9c3d83ce7034440b478d9cfaa1d080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 00:34:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d68f1fd19fed0f21ce244809090e744141684516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 00:34:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4fef06cab194464bbcf8817a8ad0098 : SUCCESS in 3m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c551cb8153648a1bf9edb121434ad2b : SUCCESS in 1h 07m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aef9ec49e5004982a8ccdcbde5eecf88 : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78616e0da22940a0b602b127a2f79209 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1228b40416472f998c8db351876602 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60282d38a94049e3a56009fcd8cc1380 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603fc93b7e1b4e2c861d0fd80aabd260 : SUCCESS in 4m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25946905f0ce4355b22c71d9c7faf3e0 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7af9bc4c107f6600f5e02d85943bd11799d6f69d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 00:41:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33955ae9508d425c82acfc23dea69198 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8dd381e626d48f11c6b8c9e296433b23e88124b","revisions":{"2e687ad78fce134e95e3856ac90365b5eb61fb11":{"kind":"REWORK","_number":1,"created":"2020-07-05 23:35:52.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/79/739379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/79/739379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/1"}}},"commit":{"parents":[{"commit":"61a0ff8ace87e917d31231f28f3be6b21a113624","subject":"Fix empty \u0027args\u0027 and \u0027kwargs\u0027 when using \u0027hide_args\u0027 with Jaeger","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/61a0ff8ace87e917d31231f28f3be6b21a113624"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 23:35:41.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 23:35:41.000000000","tz":-240},"subject":"switch to importlib.metadata to find package version","message":"switch to importlib.metadata to find package version\n\nImporting pkg_resources has a side-effect of reading all of the\nmetadata for every installed python package. The newer\nimportlib.metadata module can load the metadata for one package at a\ntime, which makes this library load more quickly and improves the\nstartup-time performance of applications that use it such as\npython-openstackclient.\n\nimportlib.metadata is part of the python 3.8 standard library and is\ndistributed separately for other versions of python.\n\nChange-Id: Ib1870a3d102116f84c7677601fd44fdac41a13a6\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2e687ad78fce134e95e3856ac90365b5eb61fb11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2e687ad78fce134e95e3856ac90365b5eb61fb11"}]},"branch":"refs/heads/master"},"e8dd381e626d48f11c6b8c9e296433b23e88124b":{"kind":"REWORK","_number":2,"created":"2020-07-06 09:07:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/79/739379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/79/739379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/79/739379/2"}}},"commit":{"parents":[{"commit":"61a0ff8ace87e917d31231f28f3be6b21a113624","subject":"Fix empty \u0027args\u0027 and \u0027kwargs\u0027 when using \u0027hide_args\u0027 with Jaeger","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/61a0ff8ace87e917d31231f28f3be6b21a113624"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2020-07-05 23:35:41.000000000","tz":-240},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-06 09:07:17.000000000","tz":120},"subject":"switch to importlib.metadata to find package version","message":"switch to importlib.metadata to find package version\n\nImporting pkg_resources has a side-effect of reading all of the\nmetadata for every installed python package. The newer\nimportlib.metadata module can load the metadata for one package at a\ntime, which makes this library load more quickly and improves the\nstartup-time performance of applications that use it such as\npython-openstackclient.\n\nimportlib.metadata is part of the python 3.8 standard library and is\ndistributed separately for other versions of python.\n\nChange-Id: Ib1870a3d102116f84c7677601fd44fdac41a13a6\nSigned-off-by: Doug Hellmann \u003cdoug@doughellmann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e8dd381e626d48f11c6b8c9e296433b23e88124b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e8dd381e626d48f11c6b8c9e296433b23e88124b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
