)]}'
{"id":"openstack%2Fosprofiler~770000","triplet_id":"openstack%2Fosprofiler~stable%2Ftrain~I8762bad8821868c97af8114e49f51b6383df3e5c","project":"openstack/osprofiler","branch":"stable/train","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"I8762bad8821868c97af8114e49f51b6383df3e5c","subject":"[stable-only] Cap bandit to 1.6.2","status":"MERGED","created":"2021-01-08 21:53:01.000000000","updated":"2021-02-03 13:21:12.000000000","submitted":"2021-02-03 13:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770000-cap-bandit-1.6.2","meta_rev_id":"6c6fee00f4bedec4fa65175c51cec130f2031c7e","_number":770000,"virtual_id_number":770000,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 13:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-16 16:23:06.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-03 11:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-03 11:46:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 23:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-16 16:23:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-03 11:46:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"71d2f5804230ff22e6a298f04b0ced4c450a8d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-08 21:53:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4205fda4684e95f45d0fdd52314de727b1f61217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-08 21:59:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1628f7732506912039e11c512b8809a5f1855c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-08 22:10:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85233833712671ec377b0bc6024fcd0b48946acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 23:46:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ae0bfe5d5704260992896d07de8b26d : SUCCESS in 3m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/75160e893ce04169af4c2bda4d1fe79e : SUCCESS in 1h 30m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d4d3661988842149979c8c6dfd207dd : SUCCESS in 1h 26m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe657b6add3c489cb7367f34b30238d1 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0a52d57916242ec86b3b8502e24bc4a : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0498c5bfe44778ba00e2e99866793a : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/452a58efc2c244888affb0daa892d65f : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e9c2cd25b6f40c18bfb10d342264b9e : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f1b9a885b7c45289071b5829887306b : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b54460586c4adf99038dcb25bde6da : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5bf96739a19451ea66655382a156041 : SUCCESS in 3m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/da4b06af99be48f18d11c6bae027f1bc : SUCCESS in 3m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de77c34cd4c2440cbe46e4190bfbf26b : SUCCESS in 4m 04s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/a3a82b97d36c4f9bbf19101256bf4d4e : SUCCESS in 34m 13s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/d9cf50196d314c729f6322efb1ff0ffe : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e57ef10d7daa4cfc712c1a3cad71a67c9cdc3341","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-16 16:23:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"023b5f9743a15352205788743b57c991e3ab93a3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:46:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"58f502346e8e95e687dd63d53523cfb93b95600c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:47:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ddee2f20aa9df521c9173745e2994f262977cd24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:18:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b38beffe4591490d87489dda7f861540 : SUCCESS in 3m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2ecb59ae50024bd582b3226c2176d44c : SUCCESS in 1h 25m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb6b8c96d1114c9eb581870cacf71631 : SUCCESS in 1h 24m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6957455d080f40c39b074e539a0215f2 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc5330585684835a3c366c5821f7130 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71633de9a59148c1a945a8c283c98410 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91ed44fb5ff94889b054937ec2424dd2 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00cf009e5feb4515b23b79078f80672f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa06c90348c4d3aa0858df61fef2b6f : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b40f859e91c4c9a83de8b2163529837 : SUCCESS in 3m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/27d0020b6eb2493486f2e21e8cb4c123 : SUCCESS in 4m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3b49ddfe4e24301aa777fb15c9dbc2d : SUCCESS in 4m 44s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/cf53cdfee1054925a00c7885c12203af : SUCCESS in 34m 16s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/0654b4de702844599504e0d0d8031cd4 : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56c3f629cace2241870d23c8ac18c4e00521804a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:18:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6c6fee00f4bedec4fa65175c51cec130f2031c7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:21:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8059b946573440581eafd3b48734c23 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f3ac9aad349495b9e45275ad78d86d0 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"30ffa278735aa0f14e80c703ccfb4252e0a4a879","revisions":{"7267ab5e303a0358f0f11f28ae257c50ed295161":{"kind":"REWORK","_number":1,"created":"2021-01-08 21:53:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/770000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/00/770000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/1"}}},"commit":{"parents":[{"commit":"2744494f607ad77508bcd84caf8b3d301a36e199","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2744494f607ad77508bcd84caf8b3d301a36e199"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 21:52:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 21:52:35.000000000","tz":60},"subject":"[stable-only] Cap bandit","message":"[stable-only] Cap bandit\n\nChange-Id: I8762bad8821868c97af8114e49f51b6383df3e5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7267ab5e303a0358f0f11f28ae257c50ed295161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7267ab5e303a0358f0f11f28ae257c50ed295161"}]},"branch":"refs/heads/stable/train"},"95e595a202af02659c8190b833bd6d7125bd5786":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-08 21:59:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/770000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/00/770000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/2"}}},"commit":{"parents":[{"commit":"2744494f607ad77508bcd84caf8b3d301a36e199","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2744494f607ad77508bcd84caf8b3d301a36e199"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 21:52:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 21:58:02.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I8762bad8821868c97af8114e49f51b6383df3e5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/95e595a202af02659c8190b833bd6d7125bd5786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/95e595a202af02659c8190b833bd6d7125bd5786"}]},"branch":"refs/heads/stable/train"},"30ffa278735aa0f14e80c703ccfb4252e0a4a879":{"kind":"REWORK","_number":3,"created":"2021-01-08 22:10:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/770000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/00/770000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/00/770000/3"}}},"commit":{"parents":[{"commit":"2744494f607ad77508bcd84caf8b3d301a36e199","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2744494f607ad77508bcd84caf8b3d301a36e199"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 21:52:35.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 22:09:55.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nsphinx requirement needed to be updated to make requirements-check job\npass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I8762bad8821868c97af8114e49f51b6383df3e5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30ffa278735aa0f14e80c703ccfb4252e0a4a879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/30ffa278735aa0f14e80c703ccfb4252e0a4a879"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
