)]}'
{"id":"openstack%2Fosprofiler~777803","triplet_id":"openstack%2Fosprofiler~stable%2Fqueens~I95461280d11b14199ba64a2da5709a1e2f4531df","project":"openstack/osprofiler","branch":"stable/queens","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"I95461280d11b14199ba64a2da5709a1e2f4531df","subject":"[stable-only] Cap bandit to 1.6.2","status":"MERGED","created":"2021-02-26 18:18:42.000000000","updated":"2021-03-02 12:04:55.000000000","submitted":"2021-03-02 12:02:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777803-cap-bandit-1.6.2","meta_rev_id":"f98676905270f5456d15b847b86cbaca67ca583e","_number":777803,"virtual_id_number":777803,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 12:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-03-02 10:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-03-01 15:57:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-03-02 10:13:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 20:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 15:57:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-02 10:13:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"82e170fc505008c9ffc9f23c8520a1649422bb5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 18:18:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29875157b4b4bcb93f57dcc4efa59bcb0b00a2e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 18:34:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af0ede2979b6aebf6440841753060f858283d5ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 18:49:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"abdc19a92d2d2d7fe75236cefa6d313738bdd632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 20:24:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21a22df999846a3afbfd9f9b056a679 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e332f1f1b9564c9499ac93438a33a796 : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7ea4b88f11546a2b47b515410e94bea : SUCCESS in 4m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9d11baa953142929163643ca745d7ff : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f185239f89485fa4e01617741f3def : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4a25991c754405a9537aad9c11a67a2 : SUCCESS in 4m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08a17a1db9274abe95eda4c37844c75b : SUCCESS in 1h 29m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b5f99feef8f44c79f53e12932ea003b : SUCCESS in 4m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f7b67a1c8d7a4f668a46c744412d864a : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8915dc4389d44ce59485bc02ba5f2fae : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"78285fbe4203aab66f88b041d35ce95bdfc78f45","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-26 22:42:05.000000000","message":"Patch Set 3:\n\nThis patch fixes py27 stable periodic job failure: http://lists.openstack.org/pipermail/openstack-stable-maint/2021-February/thread.html#30819","accounts_in_message":[],"_revision_number":3},{"id":"9adb916cb982abd7366a2db0a9922155772ef598","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 15:57:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"40ad14e86ae77bb0710452bd2b9c4ec100649100","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-02 10:13:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3f68291e7c956f5c195988d42ef45815ff339f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 10:14:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59b0ee794d9277387aa46982448c28e322510a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:02:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9bc831e30e4bcfae710d68238ad5c5 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/032bbe1e8a99499d86fbdaebf67f4494 : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e12fbe0aa1024912bc762d5a4f60735a : SUCCESS in 4m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51b949d7220b4aa3a8b5b94052281694 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7b69e8890f44eeaf74749446a4a000 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9652a10a9a97474d954e045c840f8b3f : SUCCESS in 4m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7fe7c4c5c66941da815beb7a4108da45 : SUCCESS in 1h 22m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e5f4ef2e34364efc9184cdc11a1536bd : SUCCESS in 5m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/90f22ace93484e769e2068fc1c3cab31 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":3},{"id":"37a5d6fbd498fa178ef14b502a56accac8d120b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:02:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f98676905270f5456d15b847b86cbaca67ca583e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:04:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402717615fd2478a9689266e24b74244 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ce04c95ba444189772ec42ad482168 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c3f9d3dc91cfffa86e0fc921da82c261ca0f6b21","revisions":{"1e7a457d47b4fa363233537119b5349f5e86068d":{"kind":"REWORK","_number":1,"created":"2021-02-26 18:18:42.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/777803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/03/777803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/1"}}},"commit":{"parents":[{"commit":"eff57e09b55aaf1f00b01c289526ceddefdf25d9","subject":"Use upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/eff57e09b55aaf1f00b01c289526ceddefdf25d9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 09:43:52.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-26 18:15:50.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nsphinx requirement updated to make requirements-check job pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I95461280d11b14199ba64a2da5709a1e2f4531df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1e7a457d47b4fa363233537119b5349f5e86068d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1e7a457d47b4fa363233537119b5349f5e86068d"}]},"branch":"refs/heads/stable/queens"},"de8c9078a8b52fc43b5d9faebffd067ddeef2109":{"kind":"REWORK","_number":2,"created":"2021-02-26 18:34:56.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/777803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/03/777803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/2"}}},"commit":{"parents":[{"commit":"eff57e09b55aaf1f00b01c289526ceddefdf25d9","subject":"Use upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/eff57e09b55aaf1f00b01c289526ceddefdf25d9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 09:43:52.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-26 18:34:31.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nsphinx requirement updated to make requirements-check job pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I95461280d11b14199ba64a2da5709a1e2f4531df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/de8c9078a8b52fc43b5d9faebffd067ddeef2109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/de8c9078a8b52fc43b5d9faebffd067ddeef2109"}]},"branch":"refs/heads/stable/queens"},"c3f9d3dc91cfffa86e0fc921da82c261ca0f6b21":{"kind":"REWORK","_number":3,"created":"2021-02-26 18:49:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/03/777803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/03/777803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/03/777803/3"}}},"commit":{"parents":[{"commit":"eff57e09b55aaf1f00b01c289526ceddefdf25d9","subject":"Use upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/eff57e09b55aaf1f00b01c289526ceddefdf25d9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-25 09:43:52.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-02-26 18:48:58.000000000","tz":60},"subject":"[stable-only] Cap bandit to 1.6.2","message":"[stable-only] Cap bandit to 1.6.2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] but the release\nis faulty and pip still picks it up for py2 [3][4], so cap to 1.6.2\nwhen using py2.\n\nsphinx requirement updated to make requirements-check job pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n[3] https://github.com/PyCQA/bandit/issues/663\n[4] https://github.com/PyCQA/bandit/issues/665\n\nChange-Id: I95461280d11b14199ba64a2da5709a1e2f4531df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c3f9d3dc91cfffa86e0fc921da82c261ca0f6b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c3f9d3dc91cfffa86e0fc921da82c261ca0f6b21"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
