)]}'
{"id":"openstack%2Fosprofiler~840711","triplet_id":"openstack%2Fosprofiler~master~If6de005dd8229fe6f21dc103f50db083a415d7e4","project":"openstack/osprofiler","branch":"master","topic":"broken-ci","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-02-08 10:47:52.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-02-08 10:47:52.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-02-08 10:47:52.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-06 15:26:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"If6de005dd8229fe6f21dc103f50db083a415d7e4","subject":"tox: Add functional-py38, functional-py39 envs","status":"MERGED","created":"2022-05-05 17:02:12.000000000","updated":"2023-02-08 10:50:07.000000000","submitted":"2023-02-08 10:47:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"840711-broken-ci","meta_rev_id":"db4bf1b5253674d158137a68c02f252ad2bea588","_number":840711,"virtual_id_number":840711,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 10:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"all":[{"value":1,"date":"2023-02-07 07:55:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-02-08 08:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-02-08 09:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"REVIEWER":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 18:22:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-09 13:25:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-12-11 22:10:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-01-26 21:40:20.000000000","updated_by":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"reviewer":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"state":"CC"},{"updated":"2023-02-06 09:59:11.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"}],"messages":[{"id":"6cb59d68e1a5beb0ba1333c00c2340282672aee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-05 17:02:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daed510b9e598dc71f0f360b2e073e23c38716aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 18:22:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dbfe6f9c10d43dabc6880ea604a59da : SUCCESS in 1h 18m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4288d021ec824ef698ed9f4b1083163c : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f6c12e96034777bbe742440cefac74 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b43b377aceb4ff39862d181b6df3f35 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3735ab5bb53a413eb108df587892eafd : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d6ec5c0c03b4a319edd79c636cfaca9 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6413332a86ac455d9079fe59ae9ec776 : FAILURE in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d13f2815b2948a788754f26f228f90d : SUCCESS in 4m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/114214393c2c4b4ebdc2882b3cd95bb2 : FAILURE in 3m 09s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/2af551f1191045158170cf39eae6d87e : SUCCESS in 35m 49s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/36fa8458ac6f4e018b0d7428f7860c57 : SUCCESS in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96f627eadd93a74d93aa45edecfd471dc732f737","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-09 13:25:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d6bcc988027214e6c7eefdf6d3fb6dcc89b921e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-28 21:01:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3aded589decb72d8b04cb0196039abd09179228b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 22:10:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38e3c344ae6b4fb5be6a841035e18f09\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ae4f0915f61434cb5b16e8a034e0599 : SUCCESS in 1h 07m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce9587b74f3457fa83b342f638c31cd : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6c2ba4027f4e4782dc9194045a538d : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a110a4814aee466f98913064caf1e3d6 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b66f4f113046acb42e65ab2fda0888 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bc5a6e3a57431d97a6b0b1cec4946c : FAILURE in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cedeb8e945f4d1783e66322520fcbd3 : SUCCESS in 3m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/996a5cf8d4334b46af8ea299f0214bec : FAILURE in 3m 01s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/0c516392a134412192c19dca6e18ff55 : SUCCESS in 31m 18s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/13b7b62405a64c8db97fba4663d395b4 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be51bd799902bedb55e053ffbb11f0e2a216552e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-05 10:23:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f65bd5087e8514fe489c04b06bd2c526374281a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:47:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2c9b2d1bf6a4825b974f72ac66d6ca5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e9dd38a95d344a59f4577dcdacf00b9 : SUCCESS in 1h 22m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a99c8627d0140ecb41843d12eb1113f : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a160b386264bac90ef30f2b26f96c9 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1bccb86b91043fb9ab5d1450219e25e : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdb41e766a9c4bdeb173512bb9dfc99d : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca168cce1fc64b9490f7c751ff80eff7 : FAILURE in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98cf262253714ddd88616648888fb655 : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b79a380da9414733866c6ffdceef8ca3 : FAILURE in 3m 43s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/edb58934e5df470ebf4c8f384f0ba74d : SUCCESS in 36m 02s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/d55360ae4f1c4b0b87e517f232a87f31 : SUCCESS in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c265e487d0b2a9cfb2ba6e0eaedf725112f5d692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-09 15:20:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"155dd92905a2b60205a838f0140168ce2328bb32","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-09 15:41:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5a411541b4bef5c9bab54f3c969bf0f56e19d55a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 16:51:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a8e05bcabe24f6bbe90fe92e05a4244\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8751bb0c9004f78a65edd783b5752b7 : SUCCESS in 1m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10906f64a38242fa8f5393c645e119fc : SUCCESS in 1h 29m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ca9c53062e74f7ea777b042292c109d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e783e06118f4b5e8f95e1cd6c9d91f2 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9913ccf51d094848a8f804ccccb56d39 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7be9b07e4644457b932b279fe06c2f0 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6688c31110a946069b730e6fe6ff0bfa : FAILURE in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6eb5a79d4f648dcba08f48300fc3503 : SUCCESS in 3m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92ef04339db246208c5dc0469aa4e56c : SUCCESS in 3m 31s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/de9471db8432466498cb61f1fd5a6e4f : SUCCESS in 31m 10s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/b5c7694483514d358c0c45eff09105e8 : SUCCESS in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c957ddd4c047976a99e1bcb55eb1165f0c73386","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-11 22:10:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"908bd8415712488892ca7d7aa4954ff606d3bbba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 23:42:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/637a230e2b1d4cf9b7ae95a12e9d56a2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2565541f9dd6403b9eabf8d4895704da : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a621742d05af48feb2311f42c7a49c10 : SUCCESS in 1h 31m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b584fc0a1fe64522b31345776b9dc55d : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bab99c59794db19f4aea96e6879053 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3953eca0d6cd422cbea7e2575429f7e0 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b0b8c29793a4df9929dfd9e53e4e702 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a922c2e73d94e4a872fa3d2612a6702 : FAILURE in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acac1f200ef346288a66583d44ad4af5 : SUCCESS in 2m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c24427bae094d3f92377c9cabb8c7cf : SUCCESS in 3m 43s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/060036520b214f848b55db7c4f41fef9 : SUCCESS in 22m 30s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/47d0338d7fbe476488b1ef5e4f9c0631 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d33b7974c68d1a8930495200bc14ab1c03ea185","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-26 08:28:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79f63e8e6ac8cf6c6ca61b2e5b7c52ffe76af8e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 10:11:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b8217494d3c4870a7146229540fef7f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f37e17b49bf445cafc6dba672e2f5d8 : SUCCESS in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14f9ced3d60e41c9adf52176745aac0e : SUCCESS in 1h 41m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9609b91c009d4db0b69e8c800d1ee7a7 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97483e1ba8ff4bce9530c8cb8319ad5e : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/558e2239ece44b3ea7054e75a9ad3439 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb81f431ebff48368f5f38f269e7153d : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5ff3962dff402d8c9c4d4be2c7bc58 : FAILURE in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a40f2e40c914ba6a19de82ce4ff5959 : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ad0024b388e44a80a3692f7ca8041399 : SUCCESS in 3m 58s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/c8f3e86c26be4dab8c1896a098e27705 : SUCCESS in 36m 30s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/6baa8f1635bd439bb29d1cb1646b7bc7 : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49d7c911364fbcf402f8a4d5bb84add7ac2dfd8b","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2023-01-26 21:40:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8718814064c4f0b19fdb044baab3e074910a2ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 09:58:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27a540c56cf63fa3f628e88f29e93ca1e70cf6be","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 10:01:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e2599cef7b8403ae26eb1afdb7fa97b63210cf19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 11:26:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a0c5599ed0a42a6ab9c9e8d46b941c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/272e934954674b7ba6d46eff4cfd78ce : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78b0c2ddd13e443fa94d3453faffaddc : SUCCESS in 1h 25m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/268d9006afe34696aa2c50892a7a3144 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb173bead794229bf6561b8cff8bf90 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176e6a8a10374f0a9b98de538dd09381 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af24aee3bdc4105b2e6abae04c5f0f7 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ae2a47b9f44bc3b4d1aed6ce889434 : FAILURE in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99eb3a43242547d980603d65549c36f2 : SUCCESS in 2m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32fc6f0e56d741bab27732a27a69336f : SUCCESS in 3m 33s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/bdaf312fec1b48b1bba52889024877c3 : SUCCESS in 40m 12s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/56f3d457e2884cfba1694c1a5a61fe99 : SUCCESS in 51m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cb54f3b0a6eef61a9d074d09e6d45457ea0e30bb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 12:28:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5280f80dcf7c8fd7bf2bcecdb09327de990f24d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 13:15:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9355c04f4f21c8d83297fbca3324abd1465848fd","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 13:17:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8836396ae3513504e205f63d4ca1af6dd3e77c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:21:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d08a0c8186fb40859ee1fc3805fe5465\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ecf756d19aae4fe68005e38aef581f8a : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75e725bdce224b5cb145a17720c23da4 : TIMED_OUT in 2h 04m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64da1af654fc47ff970ebe8eb835485c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae2e4bfa9434bf6ac713f79827ff689 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8335922b3f5466a9580d585f9a37ba5 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da809ad96bea4f51b197844bb0ed54eb : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beeb4e2ae6bd46a5beb510bd4bf26f40 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b34cc01a78141feaee4bdeb035beb41 : SUCCESS in 2m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de81bae1e01a4808b39cc9d2487c6d01 : SUCCESS in 3m 50s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/d7a32f0c6976475f81afc539b9622061 : SUCCESS in 56m 46s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/098f686ae8ba47a39f4517405c1f64a6 : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"07cae0167a5d292458ffa99ea7c30a888a0b3a8e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-06 15:26:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"62b55d7851a1683061cdd046da0370055e028b61","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-06 15:26:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8b66fdb0b5dcb1a3da4bc95e0b22f24da4001adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:50:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f146eacc307446b8431e7e033bdfa5a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ec250cdbeec4f0eb7e68c3c93999644 : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccefa33b5d84484080c8ad8440e7261c : SUCCESS in 1h 21m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0061e9fd00241eab9e562ae00e33104 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa186e9164c14a56817b2000a3cb97e2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/579038a005144319b88698d03905ace8 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0579df7334224add999082814f036383 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483577fe13a14682892eeef37b33bdf3 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4338bc09435c45ea82831cf90f13649f : SUCCESS in 2m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca439cdeeaae4292ac52845c65f536ca : SUCCESS in 3m 04s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/209cbd68a6ae41a894f8060c7afb35a8 : SUCCESS in 31m 14s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/299af4129437481ba669719b45899b55 : SUCCESS in 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8c803d1c33d89b19b968a83c686baa6b067eee8b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-07 07:55:50.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9e47b5927bfbab1b825b50c6da6f1ff5b8fa0e29","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-08 08:30:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"66ff779ffe4500fb693d2a3377b0105f88abd03e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-08 09:02:37.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7a49b44e2e6f6bfcb018dcf39dd4ba59a31a64eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:03:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a647e0d011f06f1394d9fed0fccec0e06e4f8b2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:47:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe30f41f94b844afaebdc069805f0ed0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2591cd3a8fda4d4f8b19516a0e9ab2aa : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e344f8c6c423464b9d2bbb04381b3f87 : SUCCESS in 1h 43m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd02f82d1ea48598c569f08ce5b5cdc : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8fbce74010e4499be20f2e1a540f582 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9a7a7e610834618acae00c8f07137d7 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0899e7e39cbb4d50848c074963c5f0e6 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140a3e96ba934f2bab27ff45b74a39c4 : SUCCESS in 2m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31414897ad6c49398cddb0ce7fc6f875 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":6},{"id":"4cc11217656fdbf2142a222466e6d1d0c5622248","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:47:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"db4bf1b5253674d158137a68c02f252ad2bea588","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:50:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d5e82c6bb24a828cd57d763a4b53c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfc9cb3f8a544ceb17c97d22dae626e : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964faba28a834d768cc4c3dce29cd4d3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fa830926b573222d88df2a0a4596fe61da9bafe6","revisions":{"a694aa4d2c2768da7d4e61787cf1fc19ba87f006":{"kind":"REWORK","_number":1,"created":"2022-05-05 17:02:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/840711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/1"}}},"commit":{"parents":[{"commit":"32863011fe90aa283425b6347123c38e5c251ac3","subject":"Make some revisions in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/32863011fe90aa283425b6347123c38e5c251ac3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a694aa4d2c2768da7d4e61787cf1fc19ba87f006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a694aa4d2c2768da7d4e61787cf1fc19ba87f006"}]},"branch":"refs/heads/master"},"e8a19104da4cbe71f544e889df193d41a2357fe9":{"kind":"REWORK","_number":2,"created":"2022-09-28 21:01:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/11/840711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/2"}}},"commit":{"parents":[{"commit":"32863011fe90aa283425b6347123c38e5c251ac3","subject":"Make some revisions in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/32863011fe90aa283425b6347123c38e5c251ac3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 21:01:53.000000000","tz":0},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e8a19104da4cbe71f544e889df193d41a2357fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e8a19104da4cbe71f544e889df193d41a2357fe9"}]},"branch":"refs/heads/master"},"a01aa0a8cdee83cfa60dbdb3175d4ca300db5b5e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-05 10:23:46.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/11/840711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/3"}}},"commit":{"parents":[{"commit":"e29b1871518f8a06cc4620f55881ce5d0be0fc76","subject":"Fix formattiing of release list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e29b1871518f8a06cc4620f55881ce5d0be0fc76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-05 10:23:44.000000000","tz":120},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a01aa0a8cdee83cfa60dbdb3175d4ca300db5b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a01aa0a8cdee83cfa60dbdb3175d4ca300db5b5e"}]},"branch":"refs/heads/master"},"05bd6c01068ed6eb709c98693d8727a7a2a7be3c":{"kind":"REWORK","_number":4,"created":"2022-12-09 15:20:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/11/840711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/4"}}},"commit":{"parents":[{"commit":"e29b1871518f8a06cc4620f55881ce5d0be0fc76","subject":"Fix formattiing of release list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e29b1871518f8a06cc4620f55881ce5d0be0fc76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-12-09 15:20:00.000000000","tz":60},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/05bd6c01068ed6eb709c98693d8727a7a2a7be3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/05bd6c01068ed6eb709c98693d8727a7a2a7be3c"}]},"branch":"refs/heads/master"},"7d75f0d3888eef10d9444a24d9ae66e06308ad68":{"kind":"REWORK","_number":5,"created":"2023-02-06 09:58:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/11/840711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/5"}}},"commit":{"parents":[{"commit":"e29b1871518f8a06cc4620f55881ce5d0be0fc76","subject":"Fix formattiing of release list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e29b1871518f8a06cc4620f55881ce5d0be0fc76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-06 09:58:33.000000000","tz":60},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7d75f0d3888eef10d9444a24d9ae66e06308ad68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7d75f0d3888eef10d9444a24d9ae66e06308ad68"}]},"branch":"refs/heads/master"},"fa830926b573222d88df2a0a4596fe61da9bafe6":{"kind":"REWORK","_number":6,"created":"2023-02-06 13:15:46.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/11/840711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/840711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/840711/6"}}},"commit":{"parents":[{"commit":"e29b1871518f8a06cc4620f55881ce5d0be0fc76","subject":"Fix formattiing of release list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/e29b1871518f8a06cc4620f55881ce5d0be0fc76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-05-05 17:01:11.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-06 13:13:57.000000000","tz":60},"subject":"tox: Add functional-py38, functional-py39 envs","message":"tox: Add functional-py38, functional-py39 envs\n\nWhile here, we also clean up some style issues in this file.\n\nFix CI issue to generated changelog with tox v4\n\nCo-Authored-By: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: If6de005dd8229fe6f21dc103f50db083a415d7e4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/fa830926b573222d88df2a0a4596fe61da9bafe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/fa830926b573222d88df2a0a4596fe61da9bafe6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
